)]}'
{"id":"openstack%2Frequirements~709152","triplet_id":"openstack%2Frequirements~master~Ia417739e51fa15b8da010e15ec9ac2497ad9c6ac","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia417739e51fa15b8da010e15ec9ac2497ad9c6ac","subject":"update constraint for os-brick to new release 3.0.0","status":"MERGED","created":"2020-02-21 14:52:54.000000000","updated":"2020-02-22 00:04:21.000000000","submitted":"2020-02-22 00:02:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709152-1582329758032-b4850ae3","meta_rev_id":"e694f5f5308d1fcb4bf14531c2c8abeacee43f89","_number":709152,"virtual_id_number":709152,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-22 00:02:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-21 14:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-21 16:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-21 16:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-02-21 16:19:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-21 14:53:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-21 16:19:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-21 16:36:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-22 00:02:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e720a1693862c93d1f378bb1a0edbe08bc16f9c9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-21 14:52:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abeca0c6f6cb77bb9c9a7ff887bcade8c5ec1c33","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-21 14:53:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1beaf28d591bd8e0462b9850310ea871fc2f1cee","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-21 16:19:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ec2c384cd3408ff8004cbd2f28481b06950bb7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 16:32:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269f10d3f25f4cf8b2dc53b51c504b77 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa407590a314324b1b78ed6a3bfbfd9 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1191a6221a148e99e92c6d2f9a3629f : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3a3b9a65ca7408fb7a66ab00934431c : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30f5a2af13db4c85820bda2a74d7e05a : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f1798e81d5fd40999d121d75ccc8777d : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e1c4df170bc849df9bba0a7c582c229e : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0422bae4bc9f487182f757949fe7f6a1 : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/76b9ff1d1b0442a7919d3df15f5ef955 : SUCCESS in 10m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/32f201e199454358a0e10c85e14fddae : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/807814b3352a4a1d93dfee16a0a4e15b : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0231c82ee87944959190b9f2ab274ff1 : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e3e15d786394fe183dd5e8230ddaf48 : SUCCESS in 25m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f6ff54ed3c54fce91f8d4d1034f3714 : SUCCESS in 28m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/13a09a55a99343e893cb54f22de7b9f4 : SUCCESS in 16m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d7fe268cee0d4a1fbd54da7299957131 : SUCCESS in 16m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/613687d405b244ad9cc53deda38909e6 : SUCCESS in 19m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7da804c206c741e183815c787cedccd5 : SUCCESS in 6m 10s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/529f9570720a404e933c3921f60ec490 : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6661130b1f849a39d413253af0ac30e : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8d577527e573469483e9022b62562c62 : SUCCESS in 10m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4d0621cb6a144d1389cf27b5ef569c20 : SUCCESS in 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcb473bdc4704b988d9fb1f12352c342 : SUCCESS in 59m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2bad3fc062684f4b8cb8712a3035ea56 : SUCCESS in 1h 12m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf11246ceb5246d88faf5c88e8df2a31 : SUCCESS in 1h 21m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/403bfedf743e411cb31a7ffbf3c01332 : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":1},{"id":"61c78e7d6b31fe437582374c455396ac5e535753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 16:33:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e38264e58ea7033e024f39de18f576108e5fae6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-21 16:36:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60a4b357d7ea6f2734e590fd220ab2215ca20fb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 19:42:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4966504ddc04213836ad0684ccb9869 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/049798ff58aa4dbca8ed8ff43e190f26 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2c566ca8ef34168bb0df169923cfcc8 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0052c9ea4dce4e898082aa96c2ae87ad : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a210525eb104a6dac2e774428e969c3 : SUCCESS in 5m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d568461cf224aa58dd2ab78376321cd : SUCCESS in 5m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bcaef7d6c16746c3bf4ad9e8c60109d3 : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/de04119f423e48ffa01216993cc8f22f : SUCCESS in 8m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/83ae4e527870463f88bd55ba34e410e3 : SUCCESS in 7m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4c5489f5b8c426e938c2f39e517fcf8 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a7a4025d0504864981343c87b845667 : SUCCESS in 8m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e8c25e40cb774d8d8797e146bb3779dd : SUCCESS in 19m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2dc8e4acddea4453901a22ec3f7fe1fe : SUCCESS in 27m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/36323246c99c4874a42e79af773b18cb : SUCCESS in 13m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d821bd9327784ea19951681995bc1966 : SUCCESS in 15m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f5f30e679c849168f7f102fe75b2002 : SUCCESS in 19m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6b926bd5032f43dd8760357a0e61eacf : SUCCESS in 6m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9fe4bbe78c6149af8801538cb37f414d : SUCCESS in 6m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df98d8339a064315988143d200d79520 : SUCCESS in 9m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8f0fd3ce5a2a4f27ba323a8362715b27 : SUCCESS in 10m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2f9209ad95734ad3a9051a35d1daddc6 : SUCCESS in 15m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb1c48df69db4877af631e949a26d617 : SUCCESS in 54m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a141efdf3e214e808a963d9c547e61eb : FAILURE in 51m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc71f4ab0d09451f9d796372944c5a5f : SUCCESS in 1h 19m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ee21bfcc2b21408ea07b092746d2fe15 : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":1},{"id":"20630e8b1f6a8cf53d34c370d8308067a207a3f8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-21 19:49:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"94ac7517fc6b5fce9812e9ef1f4bebedc9b4594b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 21:17:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5dbf1d20dbf4d50a99aca9916c32487 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec8a84a05b034590813aa33af6571a37 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4733d75c3404858bf3c7123b291a9b7 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e84d1437ae643499de7695928d1b517 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/542b7f040b8a4abe9b4eb3d3ad870f10 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef40fec1789e42c98daff951309e71fe : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7953f80b8e14ad58dae1b545e0067ec : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f65f794a75b489f910b1b1a8a6679ba : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/38ffcdd3385d4ff58c134732aded0064 : SUCCESS in 10m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a038c929f70e4a5c81e59b834121816d : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/febaf7160eab45b19aab6c5715cb16fe : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d302c48783034273af319bcd0083809f : SUCCESS in 9m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f4c90bc1c29c4bc4bbacdac2ca94736d : SUCCESS in 15m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5536742a0b2746c9bf400c410c6887f1 : SUCCESS in 25m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b04f99df4c3411ca58a9a35f062d44b : SUCCESS in 13m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4459410a9d4b4909a298ca731930b08c : SUCCESS in 13m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a3f4e90d8ea4167a009e4ef16dfa34d : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/27b024b439d344cea44988b1bcab7c60 : SUCCESS in 5m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b88df68296e44049a75ea6c75adc1705 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5581a4b4a16b4fdf86b92fbbe41f6055 : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/42bc0b5d55864083a0d432fd88952503 : SUCCESS in 11m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/29456a720b6145b7a458abd58134faf7 : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0321c88b9fa4c97b1f4d3e41185c86d : SUCCESS in 1h 01m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc21c4c0696b4f1b98c7e00288b0669c : SUCCESS in 1h 19m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32402df10d8b4d8fb5b63f494cd42d0c : SUCCESS in 1h 14m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6dcad6f9f7de43bd8b91e1f338f86a0d : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":1},{"id":"dfc7a1139e33dd5d976aef5957dde609f6de5e6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 21:18:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c09b908808252b9d8f82bd8bc5594164469866c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 00:02:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394bb73cce454b8f9949c088a9633540 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba80cbe641b4c559adc59eda55a3877 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65c6bd4cda874eac8b6d0f22ec120d5f : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ffa0bcadc694c97a45fca63a8e386d6 : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1f482ab16c6416287ef3986ae60124d : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78ec3e4468304375956b4d3ab6cdbe48 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d5efe42592c459c89a9f57aa1ecae20 : SUCCESS in 11m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7a1e879e5dbb4fdda0b036bf646d3625 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4785314abc9347d1a064f0352acb47c4 : SUCCESS in 6m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bfd5df13d30e43c59be222191c39708f : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8dd8c8b71a104fc78a7c9afa4aa64e29 : SUCCESS in 9m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4d20a3703a004c6e83baa3ef67f6dcb9 : SUCCESS in 19m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/933e2d92d3244da3a9e10c782818e006 : SUCCESS in 25m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2db66634f19448beb31b3bb641237741 : SUCCESS in 16m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/69daeff4d34b41ed96adb5115902e1fc : SUCCESS in 15m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c50b6341579c4bd6a9b14ae2ff3ff23e : SUCCESS in 17m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/42709c589a71408f91954adabbd16ac8 : SUCCESS in 6m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3977ee8eb9f248f98f63619e0f0e10d5 : SUCCESS in 7m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f38eb589ecee4a3a9bdbd65ccaf4c00e : SUCCESS in 8m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1618a04beff24e738f2c4cd617725558 : SUCCESS in 11m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8051418961c44907b5c7c60abec4c89e : SUCCESS in 13m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b2fc00b52b0419f8182f85e3655955c : SUCCESS in 49m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/485bf23c6d6242ed899adab42c3e9b85 : SUCCESS in 1h 09m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73550538c75342caa26b545dd10cccfc : SUCCESS in 1h 22m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/567c89bdabd245de8ac141fcf4765083 : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1cd55ce58245784968d61975ba3850a2a24d2db2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 00:02:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"acfee30ae1bf62155887d4aa07deb76c19030f8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 00:04:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c6055d919c49e59352acec1eb0b28f : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56ee499c87be323083d16e89ab971f4030d532ec","revisions":{"56ee499c87be323083d16e89ab971f4030d532ec":{"kind":"REWORK","_number":1,"created":"2020-02-21 14:52:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/709152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/709152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/709152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/709152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/709152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/709152/1"}}},"commit":{"parents":[{"commit":"825c615e5ae2d71f0bf0fbeaf9a5de377c09adf6","subject":"manual generate-constraints update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/825c615e5ae2d71f0bf0fbeaf9a5de377c09adf6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-21 14:52:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-21 14:52:51.000000000","tz":0},"subject":"update constraint for os-brick to new release 3.0.0","message":"update constraint for os-brick to new release 3.0.0\n\nChange-Id: Ia417739e51fa15b8da010e15ec9ac2497ad9c6ac\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: Ie9197ae5b31629b8eeff27d6b59877e0c181815c\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56ee499c87be323083d16e89ab971f4030d532ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56ee499c87be323083d16e89ab971f4030d532ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
