)]}'
{"id":"openstack%2Frequirements~709219","triplet_id":"openstack%2Frequirements~master~I9573731369a188c8cea22dd27dff0f41baca61eb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9573731369a188c8cea22dd27dff0f41baca61eb","subject":"update constraint for os-api-ref to new release 2.0.0","status":"MERGED","created":"2020-02-21 18:56:24.000000000","updated":"2020-02-22 00:24:35.000000000","submitted":"2020-02-22 00:21:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709219-1582330919160-a5ae4dff","meta_rev_id":"b65073fdaeae95f90f147df47c67ffa304ed2145","_number":709219,"virtual_id_number":709219,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-22 00:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-21 20:19:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-21 20:19:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-21 20:19:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-22 00:21:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fed399b7077b6cddba70c47bf9b90ef744082f5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-21 18:56:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f779beab3fb0e0d704ebaa2d5e2a2d1b0473ab0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 20:05:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f8a0a3b93fc48ac9a49606e9657b099 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5f9de0c9a634046ae15511161401c16 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f071e41d4be949da8d56197741fb101e : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb611b245aa84df4b440c33c152ac559 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/407a4753df004473b116661988f64317 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ecac53bff3742e9982627ddef0aa9dd : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ab9cf2c8f9e4f45b7a475215f6e45b8 : SUCCESS in 6m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65de0ecafdaa400f95aae0a2335b53fe : SUCCESS in 8m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8810acf2a1b45e5973ba3e5fd033779 : SUCCESS in 11m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e20c2625d7a64718991c818c7cc0281f : SUCCESS in 6m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ac3aa5016d0422491b542903639bbda : SUCCESS in 12m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad778dc92cbc420d9378a4b82e0c0f53 : SUCCESS in 9m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a00b1f0421ce4b0da69593ca3a071720 : SUCCESS in 20m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6cfafeed4fb044b28d1ac689cd524e16 : SUCCESS in 23m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e93a31ccd3ab4e09b283bd833ea4e7c5 : SUCCESS in 14m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/453e991df8b242a396339a0dfc485437 : SUCCESS in 19m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9abbfea361b4f138d96220d96f3bfdb : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/13dc763fbafb4b7e8fa7ae8368681ed7 : SUCCESS in 7m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4d36566ca25c4a05b17cb0a7540ab7e1 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8cc005c5023d419ca263c0ee49f5ac70 : SUCCESS in 8m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d570dc56320a4da3b1af61b14c446425 : SUCCESS in 11m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/86c3505d416b42aaa644804d1a343a39 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b76fab8135a468db4969d1413eb16a1 : SUCCESS in 54m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e949b1e32e343d790a7f3e509287e2e : SUCCESS in 1h 03m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f6cdea0d8fb42f8b9464bfb0dc05ed0 : FAILURE in 1h 00m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e27c02680d634e7d8bcdcf565c29c459 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ed91136665e66848c499b605fc6c191553757305","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-21 20:19:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"56dc19fb949cc8ea7222271e1cb930745ebeb7d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 21:48:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dfcfbfe3d954efa80d34bd3e7b7ac28 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b1d9cd14c9a4b24a202a09d30f743bd : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09930e45319b4f1683ffbe99a965f00d : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6a7d9a445a244f6a255a09ffa9c8911 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a894232ca404ab295f402c639671fd1 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2916a8dfb53e437c8467703026650915 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/822fbba9535743c8bcdee5b15aead56f : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/68f34eaa56b14f7b9609651102204184 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3aacd06ec5ad4bcaa9f98297e8d97cd7 : SUCCESS in 8m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ec810e2c8f5f4083ba8151dcae53f0e1 : SUCCESS in 6m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ca5cf53e3b2e48c1a16b82b835a411bc : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37dad6d302f14a08a2992e44519674d9 : SUCCESS in 7m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/17e0bc66e48242fb9c32ad62a7be0aed : SUCCESS in 16m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c519e4bc30a49d58523d50adc425fd4 : SUCCESS in 28m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/79b29925f06f4fdf9b5c0a605f35dce9 : SUCCESS in 15m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a2a7f3f66f4843a48698ca6664f955ed : SUCCESS in 14m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/72e1b74b3cb94ab28e5c8f0777e48525 : SUCCESS in 17m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/12b332d8382549acb125b5f34c536622 : SUCCESS in 5m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/144a9b35ef794758865753fc8b4c2526 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80da52e2456e4ca4a270b4e75b75874f : SUCCESS in 8m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3400ba35d0064f489de87506d1cbd8b4 : SUCCESS in 11m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2e722e53e2374752ad7b2e8a31464d0b : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8be1e399d6f94ebcb476438c59849451 : SUCCESS in 57m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fe7d39600d44870b1cfeb9b3f822328 : SUCCESS in 1h 16m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41801d0f042342b089a320f0e9660c4e : SUCCESS in 1h 23m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8b5abd96c0b44334a5543ee6345a768d : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"14d5702475abf4e482a1f4f0a8f9d016e9b979fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 21:48:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c71e57f428d39c37fe80da2adb214844640a466a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 00:21:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8620f257597c4eefaa552a05a98f5bee : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d8d01191c941eba393898489e4e08a : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf1c86644cbd4060ab49e1299d5f4e98 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a075378451ea4acdbb6a93e47677ece7 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45c56710f04647119547fbca90d09879 : SUCCESS in 6m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/863722ab1da945a3b5bc2b6c92af22e4 : SUCCESS in 5m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59341bf641594b2aa0ff1940e93fde1e : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f26c808123a54c8e8bb32f7e769a6052 : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7ca7c3bc5f42463592ecac7afa504246 : SUCCESS in 7m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f37ab99446384cbab28e39b687aea943 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b4e4c8328424de2bb56d436c0672bba : SUCCESS in 9m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bba61260351540e583d364f36f1dc092 : SUCCESS in 19m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/40e79171503d4c93b57e9bd20975802b : SUCCESS in 23m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/128ec10ff28045559897533e443a2786 : SUCCESS in 17m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/511d7ea25dfc4531acf27909ca0744cf : SUCCESS in 17m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/790a8144abd540329bc6dfc81762f16a : SUCCESS in 20m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d59d4b457bc44265a481a4c053491834 : SUCCESS in 6m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/583dd58f36614a378c6716b8793e1a90 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ef908509a9bf4eb0b187c07427cc3eb6 : SUCCESS in 7m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/419950cd98d1407b8c6209c425c3e923 : SUCCESS in 16m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/264f2f4fa5fc42619b15c244dc2ab6a5 : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5195a2cb6009457197d00d069f095412 : SUCCESS in 54m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf270cc01dbd4f9a9c1ae62a9401a58f : SUCCESS in 1h 09m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36fe0444f191406ba94315101df2c0bc : SUCCESS in 1h 17m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c300b562551b48218f8a74a08d75d1a4 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4513b815ea797490467cfc7726e749ce9990fc09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 00:21:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1a6b480864f7f16bee6d6e93106f83c2ae5b8b2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 00:24:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5dd11503a504cf0963a84b945ca3b80 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00769145bea7b4ee7b256c986d494edfc51ebe82","revisions":{"00769145bea7b4ee7b256c986d494edfc51ebe82":{"kind":"REWORK","_number":1,"created":"2020-02-21 18:56:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/709219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/709219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/709219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/709219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/709219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/709219/1"}}},"commit":{"parents":[{"commit":"825c615e5ae2d71f0bf0fbeaf9a5de377c09adf6","subject":"manual generate-constraints update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/825c615e5ae2d71f0bf0fbeaf9a5de377c09adf6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-21 18:56:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-21 18:56:22.000000000","tz":0},"subject":"update constraint for os-api-ref to new release 2.0.0","message":"update constraint for os-api-ref to new release 2.0.0\n\nChange-Id: I9573731369a188c8cea22dd27dff0f41baca61eb\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Commit: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Change-Id: Ica9ff1e4fa2da0343419288be303901a697985bc\nmeta:release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00769145bea7b4ee7b256c986d494edfc51ebe82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00769145bea7b4ee7b256c986d494edfc51ebe82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
