)]}'
{"id":"openstack%2Frequirements~709279","triplet_id":"openstack%2Frequirements~master~Iadc4554ba3c2d38f629ac8c80a72851aecda38d6","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Iadc4554ba3c2d38f629ac8c80a72851aecda38d6","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-02-22 06:29:14.000000000","updated":"2020-02-23 19:37:46.000000000","submitted":"2020-02-23 19:36:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709279-1582486575620-a663f06a","meta_rev_id":"104b1d6295504a6cac4e84bfa2421eac714d9dd2","_number":709279,"virtual_id_number":709279,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-23 19:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-22 20:50:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-22 20:50:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-22 10:05:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-22 20:50:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-23 19:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a737c2cb38f516d58a0eb48a532536cdbbce38d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-22 06:29:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8689af58ec3aba9b8bd64347ca2e189773278a04","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-22 06:33:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"70cb09f62c12ae3971b100ad27df7b80ec5394aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 08:08:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bae6c54dea940a29e86e92d9e417b48 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0153a8b2c514c35bc7847371e6d2f1d : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c6b161abf544fa4bd1fa66a92602a98 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2afe01bb88bd4068933b4fc74f6d51a6 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f2d3447676e42f4a7cc80606587032a : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8bdfa88a24143e7b6864df2e064aba7 : SUCCESS in 5m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd3add92dc1d4da99bd1a07bd73b1411 : SUCCESS in 6m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb5dd1369427455e8d5920b565014292 : SUCCESS in 10m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4558562f57ee4b5fb64014ee9ccfee6e : SUCCESS in 11m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/136758dbbb3d4180abcbdc0242391e2c : SUCCESS in 6m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ee8c76139aeb42099b7bc4f4a9ca950f : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c43cfafedc524f2ea30436b745357e34 : SUCCESS in 8m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3e01ebc737db4d7ca43902f45b7ccfc0 : SUCCESS in 27m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3ad27e58188a47dbaec31f54981183ea : SUCCESS in 25m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b3251cb19c4f4b5f8687fea3efc76573 : SUCCESS in 14m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/df7b94228cb54eb5be90c994b003c855 : SUCCESS in 15m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a4f413a0c9db439bbaa1823279cfe072 : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b71ee6423e47403cb7844fd92a54427c : SUCCESS in 6m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7778f3ce117d42d5a19ddd2ffa43fd27 : SUCCESS in 7m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2170c6522174dba9529e27349b912b3 : SUCCESS in 8m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bc07fc7cc9a241adb08d003fa9e5c1ba : FAILURE in 10m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fcda092bf4a2477d90e88a671da4e4d6 : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6c323134e6c4b01a288b44059301262 : SUCCESS in 59m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/efedbc8ddcbc4a69b514787afd8b1b30 : SUCCESS in 1h 11m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3cfd6790ecdc44f5a9ff0d6ddbe679eb : SUCCESS in 1h 28m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/39f523a821dd430180f431c509b0163c : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":2},{"id":"bcc979db2f7503244041d2cfb2f480877491f9a0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-22 10:05:04.000000000","message":"Patch Set 2: Code-Review+1\n\nHello,\n\nLGTM, only +1 for now to let\u0027s time to discute about the following issue.\n\nA lot of failures on import enum exist in job cross-swift-py27\n\n```\nFailure: ImportError (No module named enum) ... ERROR\n```\n\nHave you ideas about this?","accounts_in_message":[],"_revision_number":2},{"id":"07d0fd87de94e0f17e7da66cc31d6c559224db6b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-22 14:28:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"967f863a7a02e846c9fd9f6995bf117c2a7f1f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 15:45:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8636e7aba0f4972ac0de1700aa8e19b : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a7ae69a6f7f4b0493e23648203b90cf : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be94ac42102e44d985e711d4d91ef868 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d246b2a1b254ac9965fb26e7cf49d88 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55dcc302d1034b03a76f9a4a39bcdc4c : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1d6a6335ab64bdcabd0f627c04fc242 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c033b0f6aeb4f7fb0f4c014f27953e3 : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/403a99ee0c9849cebfab7de72771a508 : SUCCESS in 8m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1c4199feda444c8785b80727f090b12d : SUCCESS in 11m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/178ab7efed2b46f5af016281302699f1 : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/45ecf1617368488eadf97b0be06fed3a : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9cbbfe1eb66b4f568e9c3d43825a1049 : SUCCESS in 8m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/497fae6bbd6249d0881828fc924ba433 : SUCCESS in 18m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d40a76bdc260402a8123b4f1595bf618 : SUCCESS in 23m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a03b02a5177345e990e342051dda6838 : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0b0ace7c50bc4c56a91b49ab05c80558 : SUCCESS in 16m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8cd97c4b516a499693b37ba706b77ebe : SUCCESS in 15m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3c8691dd5aef4123aa6b5d0af2c0edb6 : SUCCESS in 5m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/46e00cd062704f98aa84856af8ae7484 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b9e68ce1cdd4d2da9164b181313dbed : SUCCESS in 7m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e9a55903ea21495aa23bfeeb94afefbf : SUCCESS in 12m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c3d5ec46f8214f7c8b2aa553551698fc : SUCCESS in 13m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8efb6b0a529a42dc8190519bf740c8f5 : SUCCESS in 53m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/73dd6670ff9845d1bee81f147dbb2a02 : SUCCESS in 58m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4c179298c104b99bdfdb10844ee07ff : SUCCESS in 1h 15m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b0981ad88e9e40bd82ff00da1b80d624 : SUCCESS in 51m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c9bc8896800ae4aeaa65907048159ff2d1c758ce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-22 20:50:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"06de23e53e8892a603765f7bda36145956fea211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 20:50:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ebf1d511ad68564720df067e14a5b8d410884e97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 22:17:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0987ca3604d465e9ac333821cc5da84 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7425df4538a24caa90401e987e6a0cff : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89afc99a51554d2b8343b272c63555ff : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e52464ffb5b4217a734adc6bdc4b71a : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f766429bcaf4142b0ac84e64927de6f : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f4f20f4ca244ab2bc39d894d9e5cac5 : SUCCESS in 5m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65bb2c6c95aa43a1abbb12b6eeb36812 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1aab6bb25a9b4528b53c5ade8368f395 : SUCCESS in 11m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2552270721a14febb467c17ca6e85f13 : SUCCESS in 6m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e47caab370a4401d86ed727ebaebe2a9 : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3559372bf36a4a768c4dcb0f75326eda : SUCCESS in 8m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/940aaabd494b4d56bbd2c099d6f2a77c : SUCCESS in 18m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/33966930157a427d9eedd4d716857a3a : SUCCESS in 26m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3329c45a69054b47a5daaa8feb0c741a : SUCCESS in 16m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a0b7943a018449d5b5b9ac94d54a2054 : SUCCESS in 17m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8552eb4830741068a167caa3812e838 : SUCCESS in 19m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/592dd85f536e4eb8825ea0955fe1ed89 : SUCCESS in 5m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/72157127f354409597b96058eddc9d46 : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ee61ceb04604308997ff65e83cdf32d : SUCCESS in 7m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a7e2e199d6f04f54aef8aabdcf29503e : SUCCESS in 13m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/144dc10cdbd84b3c87787d8aa8e58f53 : SUCCESS in 16m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/308fa58b12b14e7a8f77c3de2efc5243 : SUCCESS in 1h 07m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/64d7eebf7a81454b9f6b64788a01aeb8 : SUCCESS in 1h 00m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6e8f17467974e438a88f768c6adf36f : FAILURE in 1h 19m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e3640770dd374885be3c656ec9db29bc : FAILURE in 50m 27s","accounts_in_message":[],"_revision_number":3},{"id":"f8e1850f979326e7b76967937b638903e78cd0c0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-22 23:18:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ad4d96ddff7275edab5f70a5a6d976ec12c1a7fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 00:34:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21ee81323bf49b4b9b36e21a7b6c2ff : RETRY_LIMIT in 1s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5fac7f761dd4219852b5228d25f5eb7 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fc6a240e2034ba2a44f83c505ed40da : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59a12364663444b8ab4ee178928d0e2c : RETRY_LIMIT in 1s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46cc27fe49694231984b2680bd937c67 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8905de9b7f4b48d9a0cdbd30666768fc : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2abe5dc25e994094943a923cd3273b0d : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/57e4e03eccf540c3802a7b95aba8dacc : SUCCESS in 8m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ba339310488d49ae9bf4ae2603a4d262 : SUCCESS in 8m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8464d5eef4514b3ebf88fb7c9499cabc : SUCCESS in 5m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0a6b7c313d8c46e191dda21d7b18a3c6 : RETRY_LIMIT in 4s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34501c007a9544668e0a8cf8e2845b49 : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c663d8ad3af1490b95c1a21820706ee9 : SUCCESS in 14m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/940ef2c88b344bba9e1e025a7b768966 : SUCCESS in 23m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b21342595d74a9a9856052a906fbbe9 : RETRY_LIMIT in 5s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b378c6d043bd42fc8d216ee51c635205 : SUCCESS in 13m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/105541e6a48947ba888ef57282cf9d57 : SUCCESS in 15m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b8d4f5bb90643ef806b9d2d198e31fd : SUCCESS in 5m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3c3d7d186c4c444cb2cab3b4f84d4c7c : RETRY_LIMIT in 1s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b64b987e192b4b6bb3b190d1d5f3e9cc : SUCCESS in 6m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6e8bfd07800549e6acb0b218c698be09 : SUCCESS in 13m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6de3bc0ed59c43d49996bc9e55fe165d : SUCCESS in 13m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe51b844fba041988f760a939862d4a9 : RETRY_LIMIT in 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7029c05763b40b8bc5aff6b8c19700f : SUCCESS in 1h 01m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29a80019a42f4eecbb9d9a234282a3a6 : SUCCESS in 1h 14m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8276312ffcf34b1a84b2aae036680fcf : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":3},{"id":"d6177026243b01f90cfb2ef9d94b327f6b40f4e1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-23 01:20:25.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"0aa40ae7eaa6d80ded6493f113c8bfbadb5bd1a1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-23 01:20:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c38f406dd71b9ae4c99b33dcca596f954427e236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 02:51:25.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d401345ba91f46e7855f9365898d8fe0 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad194645f1cb457e929080e22e501fb7 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/355415e3800243fd910e8d654da5f5da : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0807d28994a46d08244c85749634794 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c20782f6e40248c9bb6d354cace750b0 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0257004345084660b36192e1644816e7 : RETRY_LIMIT in 2s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5b9387f300c4579a44b8f920ee85d97 : RETRY_LIMIT in 1s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c3a2669bee7e4f55917b3e407ca4502b : SUCCESS in 10m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/978e6ebde70e4b5c82d3f3d5165e46e6 : RETRY_LIMIT in 2s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f6a5101ad740435785bd189d94a068f3 : SUCCESS in 6m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9a4a245698ce46389c314f53994562f1 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9992b90422fc46b0975d8d8249bf6936 : SUCCESS in 8m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3985b3c60dba49dbbc6d85b319e8ae06 : SUCCESS in 15m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e50cc0b40469498490efc514afd6a9cc : SUCCESS in 23m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b45c01274f6b477a9a75d73ba7273dc2 : SUCCESS in 12m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bcc04bda20fd4aadacf0047127f8004e : SUCCESS in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63f53122bcc341a18cfcfb02efbc85b8 : SUCCESS in 15m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aec7c28555dc4ef49c5c4717c838f361 : SUCCESS in 6m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a712ab245b6746189b033bbd0c3f0a32 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3eb62b6635e0439fbfa40fb26886064d : SUCCESS in 8m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4bf32175ea9941f1b0dff15d2f665a3a : SUCCESS in 9m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1dc782c711ee4defa0f141174cbc48f3 : SUCCESS in 10m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/658e6f46ed094ad3a98446dc62ecb71e : SUCCESS in 51m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9f07945c8fa4de59d9a37639fc227ac : SUCCESS in 1h 05m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d58bc0ee61b8428ab1f44ba044c24a0e : SUCCESS in 1h 22m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d43ec50e1049478fa54ae51d17f1d93e : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":3},{"id":"00f325e6243d1db13e628475f21dfba5660d4c96","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-23 03:21:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"96297278b436e63df710d2c738e59a52063734bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 04:39:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d290e6daefde4dbbb48b679cef1a31e7 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/470a1947bac04c0f896ed1f118cb8d5f : RETRY_LIMIT in 1s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88ecf1b052bd4ea98432d5c9a1c41c2e : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4b3668f86f24d5a9e1838939f12bdd8 : RETRY_LIMIT in 1s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f427ae2852cc4c3e9424a58a3389d083 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1db127667f9c4438a5ad1a8b16f0a49e : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ac31fd11f90413492cb832bdc786145 : SUCCESS in 3m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15450b2458814f6db75e3dedf844d575 : SUCCESS in 7m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6702215391a548ceace86e3903ae68d4 : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/381c00e181244354a72a97a501817e7a : RETRY_LIMIT in 1s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/161fd9686ad84b05a9b8df2423238554 : RETRY_LIMIT in 3s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d32f1f1c041467dbe2dccc1b87697c6 : SUCCESS in 7m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ccabfdf0f0784ce19a35c8da9410a34c : RETRY_LIMIT in 2s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e2042fa039164fee80624c8b0c18990d : RETRY_LIMIT in 4s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/493a2e03414941c5a5e58d1dbc1050c3 : RETRY_LIMIT in 6s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9bde560865e54edfbb210a4afe6e6e0b : RETRY_LIMIT in 5s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6bd4f9d05e194175b29e1ea87e350838 : RETRY_LIMIT in 4s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1267e58f16054669b058c5364044aba7 : SUCCESS in 4m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/96efb864e6d043e984a4da5e99192891 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0d9bfc1bd3a84ca2acdc07637407bcf1 : RETRY_LIMIT in 2s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bf7c6bc5e729478ebc1fe749a4f84a67 : SUCCESS in 10m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73cc938b07804e44a4447778f53444df : SUCCESS in 9m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de1c66e8fad84fba8a40a9d25cab4c9a : RETRY_LIMIT in 9s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5942adfec9174d5983cbcfdbd35f17bf : SUCCESS in 1h 11m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/740842aaf68646f09f1fc388054b9c0d : RETRY_LIMIT in 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1a207dd4a8934e0a911c554b36340049 : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":3},{"id":"f6619c90437059bca118e4f3d28f247486369317","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-23 05:24:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fd28ce63d7ad7ad43c4ffedfa96669e57d44947b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 06:35:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7805782e64c40a7914f64b4c7334eac : RETRY_LIMIT in 1s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e86f03d706344c1eb5a17e07cfc5c2fb : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdf337bb900546438e6af59f810f7672 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce5115b950274d8395f4ffb2b28bf230 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9225f4f14aa542d2bbe0ba3a4aedd467 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/966f866e44634afa9b237d7c81b43a9e : RETRY_LIMIT in 1s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d60ebce6a7c94fefa126222f5964f781 : RETRY_LIMIT in 1s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a6bb290c67d45678a2b08e808dca677 : SUCCESS in 7m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b6fdda3dc56a415abeef80e27dc5daeb : RETRY_LIMIT in 2s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/389d54e59f9a4790a6fc14c3700c5efc : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9a64e836c45742d0acb86c20db9b1a04 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4bef21438fd447d5b4626795a356be69 : SUCCESS in 7m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b9b0ee9b0bbd44cc9ebf13b6aa78f1d1 : SUCCESS in 13m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ef372b30185d4d0ab96ae2398eb7b342 : RETRY_LIMIT in 3s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c307c01377ea4649a5e79b28a40553c9 : SUCCESS in 11m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b8a1db668fc74126a976887b37df7dfe : SUCCESS in 12m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f90d72eb126b42189158ff9466f9986d : SUCCESS in 16m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/54c646b27835483b80d9499fb997fda8 : SUCCESS in 4m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9ed78df9e70a469cbdf5a74ae1356f3b : RETRY_LIMIT in 2s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae3a9b247ff045d288c68b5cc57c7d2f : SUCCESS in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3a1df18d01394dd78f0c25516cac6672 : SUCCESS in 9m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a48938e0c3ee477b9b1e98dfb5de4e08 : SUCCESS in 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/517b96615641414ab65efe27302dcbce : SUCCESS in 57m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b624fdde0524cafbd96041e28c98152 : SUCCESS in 1h 02m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b13ab04fe7f46f6adf9e0b7d3a8c79b : RETRY_LIMIT in 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/97d3aadcc37641d7ad535be2a4298c92 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":3},{"id":"234f9123e83da7575711f5a98f9b077bf77dc984","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-23 16:51:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c30aaa539d87dc980d519277fe64fc1379c44961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 18:10:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db249e308e74d6396fb31b29983ab6b : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1671077fdc394a729e1e133ac38bc447 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9fd205380474d2cae0b3b0eed5c3b54 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a0c66f609b54f7da409dc72049cd86d : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8750784929d948c4806e1c79e51ef471 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0372c5f4bcd5487183cd7bd45296dc88 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97d600690c624d90b08e6e5cbe6256cf : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4069d1e21dca4b42a8a5afb46972fd3b : SUCCESS in 8m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/38afa4897b5b4c68abf11eee93165d10 : SUCCESS in 9m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5c7b532bd4114f73bd6706a703a8882b : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/253767b1159246f383a46443cfcf5735 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01f4030e214a43a7985331377950ff45 : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4b35c7843b134764a8eb06b30fdca4ff : SUCCESS in 19m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4f8b4b1ddd1248a789f388bbbff2348d : SUCCESS in 26m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e485764284b4d46b6bbf51c3fec3bfd : SUCCESS in 20m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a5377df421ce484fa6b5e5ae866373a4 : SUCCESS in 21m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1da36020706d4be49ad8ea7ec634c9b4 : SUCCESS in 22m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/698fefe4af7849bcae25c2063cdc2ab2 : SUCCESS in 5m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b34ac14069be415e8f971956eac8e203 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/75a6d9960eac49798afcaa6ba5a5fc2d : SUCCESS in 10m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7ddabe3b0da94831a7399769abb2d669 : SUCCESS in 12m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c4eddbe912b041989ca78e49400624e7 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7c797b279974790b3cc89d53fa373b3 : SUCCESS in 1h 02m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f139bc5ffbc3450d95f1c0a73c3a9f2d : SUCCESS in 1h 15m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b0ed1556d31409b81ff4473a975e074 : SUCCESS in 1h 15m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/99e56f92127844ad899fabbdf37471c8 : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":3},{"id":"164338e80db72d45b4481cf176cb89ef0639e0f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 18:10:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1d72765fe2fb6f4eee0d98312b63a11987766675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 19:36:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee7bd7218444603a8c9baebb5ade717 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2468b40b22b74d45bbcdecd49b2886bb : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7b18c62faa44565914e1e36c657c2ff : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/109c68416241496da154004eb20a227c : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30d78f0c34a14c1d839405356dbbfa52 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9cdd2cd8bb12470e82954f838c891ceb : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1202f7e1480c4380b6208a8880f7123f : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f34afbf9d9bb4669828821ebed243de2 : SUCCESS in 10m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5650d2db05ec411c858590d7de5dc5f2 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/67a4862084144840adfb79abaeaefc08 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4895e08978824242b67d947d9896f96f : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f31ff39c5cb46988474d350a2919f86 : SUCCESS in 19m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f079b44f971447b4a0a3eeeea35ebb22 : SUCCESS in 30m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/389b53b56a2b444ea264cdce04e60271 : SUCCESS in 14m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a8fcadbebbda487599c7f16af429458e : SUCCESS in 14m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75d4afc4ff1c430eb13d57afba1efdad : SUCCESS in 16m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c78441fb5a20427d9b97dae2cc8e8a08 : SUCCESS in 5m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e4f4848581be4206bae4168853bb186c : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5eb5c1e268664561ac668683fff90bb9 : SUCCESS in 7m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6b6178b05d474f2ba8e247a5dbac99f6 : SUCCESS in 9m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8db975439cbb4a048519b3f6fee16a1e : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba73a1a3160f49fa8671254c5b323c24 : SUCCESS in 56m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/956778507303441a80f65620c252d1f9 : SUCCESS in 1h 11m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e4ded5afeb6465f883bc5be0920e282 : SUCCESS in 1h 25m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f04e06ed2d1b4368bf3af9f6123e7dae : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":3},{"id":"299c106d4f64e417493e9a18844985dc67b4b139","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 19:36:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"104b1d6295504a6cac4e84bfa2421eac714d9dd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 19:37:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115a0e710032432bac0e07e5ad521c72 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c15e4f22eefd91eefdd53d0eecf7023999bed3bb","revisions":{"348165a68aef00a027d714f5fd71a91ebf544076":{"kind":"REWORK","_number":1,"created":"2020-02-22 06:29:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/709279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/709279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/709279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/709279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/709279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/709279/1"}}},"commit":{"parents":[{"commit":"43ecce2e73f321fc20c548c2d5c9a38aeb8cd9a0","subject":"update tinyrpc to 1.0.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43ecce2e73f321fc20c548c2d5c9a38aeb8cd9a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-22 06:29:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-22 06:29:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iadc4554ba3c2d38f629ac8c80a72851aecda38d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/348165a68aef00a027d714f5fd71a91ebf544076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/348165a68aef00a027d714f5fd71a91ebf544076"}]},"branch":"refs/heads/master"},"43c128fb80a5f8520a79aa5947c78f171a431d0f":{"kind":"REWORK","_number":2,"created":"2020-02-22 06:33:24.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/79/709279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/709279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/709279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/709279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/709279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/709279/2"}}},"commit":{"parents":[{"commit":"43ecce2e73f321fc20c548c2d5c9a38aeb8cd9a0","subject":"update tinyrpc to 1.0.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43ecce2e73f321fc20c548c2d5c9a38aeb8cd9a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-22 06:29:12.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-22 06:31:54.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nhold back docutils because boto likes to cap things\nhold back msgpack because salt likes to cap things\nhold back alembic because it breaks nova tests\n  - https://review.opendev.org/708707\n\nChange-Id: Iadc4554ba3c2d38f629ac8c80a72851aecda38d6\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43c128fb80a5f8520a79aa5947c78f171a431d0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43c128fb80a5f8520a79aa5947c78f171a431d0f"}]},"branch":"refs/heads/master"},"c15e4f22eefd91eefdd53d0eecf7023999bed3bb":{"kind":"REWORK","_number":3,"created":"2020-02-22 14:28:18.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/79/709279/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/709279/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/709279/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/709279/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/709279/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/709279/3"}}},"commit":{"parents":[{"commit":"43ecce2e73f321fc20c548c2d5c9a38aeb8cd9a0","subject":"update tinyrpc to 1.0.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43ecce2e73f321fc20c548c2d5c9a38aeb8cd9a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-22 06:29:12.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-22 14:26:20.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nhold back docutils because boto likes to cap things\nhold back msgpack because salt likes to cap things\nhold back alembic because it breaks nova tests\n  - https://review.opendev.org/708707\nhold back enum34 for https://bitbucket.org/stoneleaf/enum34/issues/27/enum34-118-broken\n  - will need to be masked\n\nChange-Id: Iadc4554ba3c2d38f629ac8c80a72851aecda38d6\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c15e4f22eefd91eefdd53d0eecf7023999bed3bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c15e4f22eefd91eefdd53d0eecf7023999bed3bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
