)]}'
{"id":"openstack%2Frequirements~709590","triplet_id":"openstack%2Frequirements~master~Ie54ebacf5378098f02b114222f33efa71c65131c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie54ebacf5378098f02b114222f33efa71c65131c","subject":"update constraint for cliff to new release 3.0.0","status":"MERGED","created":"2020-02-24 19:22:24.000000000","updated":"2020-02-26 21:28:21.000000000","submitted":"2020-02-26 21:26:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709590-1582752366211-d82724f5","meta_rev_id":"7ae05f5bd35f4e19f69ce64a4ef95c7cdc383151","_number":709590,"virtual_id_number":709590,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 21:26:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-26 19:48:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-26 19:48:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 19:48:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-26 21:26:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"460d0fe0c3136ad84fe0d68ae55288d3bc7389c6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 19:22:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a9823a64f343f54950514046ba3ff0e29e792ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 21:06:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d38ab313d8e42628b52b9438180b073 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd406bf820414695a418dddcc0bb4a2f : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c08f4b28ad774e148bb15632965b5dd1 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d52672ad2c54ce495e531e9ece1d812 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba0de91e5d974daf8af2152a188084ad : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d4d4ddcc082473ca711279729149603 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1fed60f7c534554a1a6b2f0eecdc43e : SUCCESS in 6m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e2835b12cb14a3582ccbb442d56f12f : SUCCESS in 12m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c6bae938c4714bf683b059d2b5df2ae4 : SUCCESS in 11m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8fb71b27dd864174a052a81cd723258b : SUCCESS in 5m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f57203b3f7548138b7392551b20009e : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73bf4f015397484f9f9076a0b442c7ce : SUCCESS in 10m 17s (non-voting)\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3aa82d5baead473da4a8788817029b73 : SUCCESS in 16m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ffd42a9210094bb5b850c0f918b4a7a5 : SUCCESS in 28m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0070913920f1489e917f539d1804c2a3 : SUCCESS in 15m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/719aa8d101784c0abf6b7c630442bc85 : SUCCESS in 15m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6535349a7c9f4b32be6349a8febc1eee : SUCCESS in 19m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/02c1e19bf1c14cd08598d988b304c4ea : SUCCESS in 9m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/87dc36c6cbcf4deab62539d98fb868e7 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ed6a4e22f0cd4eb689f637a035bbe5aa : SUCCESS in 8m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/577f244a008542e0821f521cace8b235 : FAILURE in 5m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b1e0e66cfb1140868a9772d74adcfa7c : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb3d20a98bc64107bc31a46f28a6b746 : SUCCESS in 1h 04m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98c541d442984a57a108fa8e6e097178 : SUCCESS in 1h 08m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2118bc79b9cd4586b8c1475454511dcb : SUCCESS in 1h 23m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/453fbfe1f44744ac988dfb2e879c0dce : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3ef1a08ff42818d747510108e6df8e6614731189","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-25 06:18:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83d05f6b87661bc7d6278ded6be3de0b2c1476c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 08:24:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d2c1a53018b420d981e97b121c50699 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1bdd810c85d4f9c97a644bc8b3915c6 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d99c7f49eaa84595b3855c5bb50cfc59 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ec8d849cf814818b84d40f28d3b967a : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f85aadda5de441cb45fdcff0fefc1a5 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad44ae091bf44e00b7e2b2f552846f31 : SUCCESS in 5m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a1a712af100408cb596ba5d7cea0289 : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/560e63f10c2344e5801da6d7ec7423c2 : SUCCESS in 12m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/33c0e80308ce45fe96fe88f08894b0fe : SUCCESS in 11m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fe01ca3f57d44a4bb3209ff39cf7136c : SUCCESS in 6m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5269eceb7784c28b7c0a840be53de54 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8fbd7d3295c480f89fa2e0ef07f6535 : SUCCESS in 12m 11s (non-voting)\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/504d6ace73524353877a84f69b146809 : SUCCESS in 20m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/838eec6298c3485083af1e54091c727b : SUCCESS in 24m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cb5b6aff0e134f5fa020a7e78846bbfb : SUCCESS in 15m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c5013c3b6ce6459a83483ac962051f7f : SUCCESS in 16m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78481eda2dac4ccf82c641416eb45716 : SUCCESS in 19m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/23a6138bc6404b50bfa0e1c7bd81449c : SUCCESS in 6m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7c15d4b507d74e9285e09fe484e54c28 : SUCCESS in 8m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8db1c0efbca04ab8b4c19dbfafb7e79d : SUCCESS in 10m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9901938af1b94320b4f4c9815d0e095d : FAILURE in 6m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a4dd6418eefd462b851d4803f8f32ddd : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18405652737344d486c50095e5150c92 : SUCCESS in 1h 01m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0957c3b0c44345f0a1584f8af84c5977 : SUCCESS in 1h 10m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa21e9c2870b4778a1b18279cad6a7ec : SUCCESS in 1h 18m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/59b723a8d0854bfa899b24f9b4cceeb9 : SUCCESS in 53m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5269a1d3b76ea708ebd89383b763aa5726e23182","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-25 16:27:33.000000000","message":"Patch Set 1:\n\nwill need to cap reqs for swift","accounts_in_message":[],"_revision_number":1},{"id":"8c1e55f03a4311072a1292e63def95fc8ed581ed","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 16:05:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"598c5f0d33f42b515e9fe37d1a8ab5059e9dd3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:47:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f904b6c7242e45b5bd44374f3c78d736 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59451cf12e1944e88f2b83ec7ffa1bb4 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eb7aa08643c4b8581f8e2b0079c41e6 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe371b57073343a1bffa4a9ac28ee152 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f46ad743c3ea49fd9e8e421f78caa9e2 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/390a017bd54946b6975c0620ad6af99e : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/960cc3111bf4429ebf9f6caaad84f99f : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3391d832a9f94f1cb2fa4a9bd3f715c0 : SUCCESS in 12m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0837105af98946a5871fa6f20d0fbd67 : SUCCESS in 11m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/20aaf4415a404e20998b712147493465 : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/115cbe39bc6245749d71e816d30b26df : SUCCESS in 13m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b59c21704b5429a89c2a0b0053eb9f3 : SUCCESS in 7m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/09b58c20ff574bde88f9487ba61de9f1 : SUCCESS in 14m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66b1d1c5ed8741fb883e988e6eb7161f : SUCCESS in 29m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d5eead840253421bbdd5def181ebd45a : SUCCESS in 15m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8cf9f6c4ea2741ac9a78df46ad6ce241 : SUCCESS in 16m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d07d708715ea48dfabb5f535f19fef8f : SUCCESS in 20m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/195dd8b13aaa44c5b15103665c56c97e : SUCCESS in 5m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5e4e96f7c6c346ef8ee777dac4cd7534 : SUCCESS in 8m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f8ff7de1dbf44e4a2d61c69e96155ff : SUCCESS in 11m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ec7dd55d0b754ed5aba05c518a4c3c16 : SUCCESS in 11m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/37587ddbadea4f8b882d964d5ae14c41 : SUCCESS in 16m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a50cd970c38b48a2b7fe790d2386d496 : SUCCESS in 1h 19m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c75a86deaad4f5c8d7a1845cd960f1a : SUCCESS in 1h 16m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b006af9b9b5b431c896bc3b5cb0c6d5e : SUCCESS in 1h 21m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/327ce5607ebd4cef838a84c087388690 : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":2},{"id":"282f7af71969f522d1bf57a6dfb21f0e4e9db8a5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 19:48:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e0a7617eaceef50f7067b2257cfffd046d4dacc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:48:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1af257373945ddfba6de1a7814ccb96d0e93efef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 21:26:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"151af043c054b84dd814bb0c835f119d2f4502e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 21:26:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e452298eac04293af30e35e9f5a17d3 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0b071b099a8458c81395000d9cbd363 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb1b5aa9f73c404d97faff8e6593bf45 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0519fdf27e84af69bddb48a8f27c5f9 : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ea44534f74d4c7db835665bb7d84120 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ce44f01261a4b43b27aff7a1d0294ba : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f79cf547b040441cba3d47469bb6b2ce : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b0249ef0fb9454480d5875850ad4e11 : SUCCESS in 10m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1ba44fafe2024a0fa8d8f594b02ff442 : SUCCESS in 7m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a4010834237a4341bac1b48701dbd3fc : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f97257ba29e42aba7e1df01f6611988 : SUCCESS in 8m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8793e41c4fc4489fa005281c3fec5314 : SUCCESS in 14m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/adbef1c02f0a47ae868f36fa81c13f71 : SUCCESS in 27m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ef1edcb2fdf14fd19a148f27596efe1a : SUCCESS in 13m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3e676898c51e44409342b765d83325a3 : SUCCESS in 17m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/119ddb5005894d1eb579f3a5b45e373b : SUCCESS in 16m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3203f8b98f6d4b82a8946035e95ce1db : SUCCESS in 5m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e69f6f4cfb754fb89da5288f879ce5bf : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/25a2f856950b45ab8bf60cc1aaf7cb4c : SUCCESS in 11m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/df7ae49577134aa6b320f9d3159d2bb7 : SUCCESS in 11m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b7c931c643344a7e9ac563dc2f2daabf : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e7eb24dc8d84af49042982b4c5bd379 : SUCCESS in 57m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d0310d036a14f199766262f324474a3 : SUCCESS in 1h 28m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cadbd27f474c4db5aa38d4f26d474256 : SUCCESS in 1h 16m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2a66908136ae47f882ca0863494e28f3 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":2},{"id":"9ca8c2e2ec84eb9f910011ed6064a1e6ccb255a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 21:28:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522b4665813c41b687870a6a14432053 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30e0250f39c4633ef084527611224e51bb404183","revisions":{"bfed5f2134df8bcb08e4d8377787fdc1c2fb7666":{"kind":"REWORK","_number":1,"created":"2020-02-24 19:22:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/709590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/709590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/709590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/709590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/709590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/709590/1"}}},"commit":{"parents":[{"commit":"7d5bea2164d74471c3cff072f73a45212b3793e9","subject":"Update gitdb to 4.0.1, gitdb2 to 4.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d5bea2164d74471c3cff072f73a45212b3793e9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 19:22:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 19:22:22.000000000","tz":0},"subject":"update constraint for cliff to new release 3.0.0","message":"update constraint for cliff to new release 3.0.0\n\nChange-Id: Ie54ebacf5378098f02b114222f33efa71c65131c\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I721f3f2533e339df8b536277278d10f6981b2aee\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfed5f2134df8bcb08e4d8377787fdc1c2fb7666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfed5f2134df8bcb08e4d8377787fdc1c2fb7666"}]},"branch":"refs/heads/master"},"30e0250f39c4633ef084527611224e51bb404183":{"kind":"REWORK","_number":2,"created":"2020-02-26 16:05:19.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/90/709590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/709590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/709590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/709590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/709590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/709590/2"}}},"commit":{"parents":[{"commit":"7d5bea2164d74471c3cff072f73a45212b3793e9","subject":"Update gitdb to 4.0.1, gitdb2 to 4.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d5bea2164d74471c3cff072f73a45212b3793e9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 19:22:22.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 16:05:16.000000000","tz":-360},"subject":"update constraint for cliff to new release 3.0.0","message":"update constraint for cliff to new release 3.0.0\n\nChange-Id: Ie54ebacf5378098f02b114222f33efa71c65131c\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I721f3f2533e339df8b536277278d10f6981b2aee\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30e0250f39c4633ef084527611224e51bb404183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30e0250f39c4633ef084527611224e51bb404183"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
