)]}'
{"id":"openstack%2Frequirements~709591","triplet_id":"openstack%2Frequirements~master~I6d1e1316e416fdaf7c94a7d04fb08fc5cd7d6ed7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6d1e1316e416fdaf7c94a7d04fb08fc5cd7d6ed7","subject":"update constraint for python-openstackclient to new release 5.0.0","status":"MERGED","created":"2020-02-24 19:23:01.000000000","updated":"2020-02-25 05:32:57.000000000","submitted":"2020-02-25 05:31:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709591-1582608698379-b16e9fe7","meta_rev_id":"e1f92a992c9fce4422481dfa8af82dc51183f333","_number":709591,"virtual_id_number":709591,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 05:31:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-25 00:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-25 00:48:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 00:48:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-25 05:31:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48117e0eba5336860139621e64eaa9c6cb6bc8ae","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 19:23:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94f0158f32122922bfe1054756f31c900fbd7392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 21:03:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4bc22379e854d4890c0c9f68e3ded07 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b49b1f806142a2a2bb6af2035b2e6f : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce7330e255164fbf8ddd9d6a79df89d6 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dba6baa5195498f92344d98eb6ae6b4 : SUCCESS in 12m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/221df2aa5c4f4a6fb5cdc429378a34c5 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eddde362016f41e39f0f2192c4885ab3 : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85c9e504646649c38c603196b84e1879 : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/582d5c0f390f45658c6f3783b8e4d295 : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/654e931d40dd4d3292e54addd0a22698 : SUCCESS in 11m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/849f7bbc16ef4e059c31c935a100d984 : SUCCESS in 8m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1ff65c7e941e432bab1f4c5bb0a0430e : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42c92ba4d4814520bdf9844ac7c23295 : SUCCESS in 8m 41s (non-voting)\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/83288b7586d94510a6e67173b17cfc5d : SUCCESS in 15m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7a09f76597374936a6f7eada17eec84a : SUCCESS in 31m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/898408a19a424d10a8c2fe2f46559024 : SUCCESS in 14m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/37315a824b574ba295709fc97f93cfd9 : SUCCESS in 18m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18cfafd3c1e74e1da90810e2cb766c6b : SUCCESS in 16m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/356112f1b9a844f2b124cecb06dd2ad7 : SUCCESS in 7m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8e5edea425e641a1841e4cb7dd48b7b5 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea5b82a31a494c37803135c7bda82667 : SUCCESS in 8m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2b2f56203d074b06ab1a872a81891eed : SUCCESS in 11m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5f52669e5e1a4b118797159dceb77073 : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b31960d88ac44629a1387145e21d3e7 : SUCCESS in 53m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d0c7d617f294ccb877e2cd122399aba : SUCCESS in 1h 05m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f49bb1e06a0f476cb0b9c93baf33e7c8 : SUCCESS in 1h 17m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/74b48d4747654e008877b7cabd0133bf : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ac8bd3732e365d635161814d87607fcbb5af6a52","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-25 00:48:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"226049c15aaade720df7701734a7438f9a92506d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 00:48:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"707ae61ec86ab55962672ef3f990722fcd9f1919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 05:31:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19297507c90e4a2c9b1484f290a7405b : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7a2ca0f89c54b199eae46f92d7f9050 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/641d84d479b849dabc85aaf84c6d463d : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22bd423d860e4ac18b4b0de761be1aa1 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c489f602c4cb449f8e61394e7c2636b8 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/632b2231e3344a579a2aac950ef9d5ad : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/221225c99b15482fb882cbb1452032c6 : SUCCESS in 8m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4923902879234764b9a8ac0a5d9f2566 : SUCCESS in 8m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8ca605d9c89a4b6a85a68f2e5e43b656 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c83e2dfb02dd4215ace42cc85e61ad75 : SUCCESS in 11m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f2f21d903c2446cc830520ce8255d6fe : SUCCESS in 13m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8b1297b134804cb48bacc0c36c9330b7 : SUCCESS in 24m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1366fb2674d54fa98faafd788030bcf6 : SUCCESS in 13m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9f0943f78147406baaf70e10b314d7be : SUCCESS in 15m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0429e1617ab64b57990838ce3461afa4 : SUCCESS in 16m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2ae8a5ec615849839e9876b17e3ac660 : SUCCESS in 5m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/69f50b7ad2504d5f8e85a216e9b8e4a4 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/145c6e98598b4278a6a6c9202813baa8 : SUCCESS in 8m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6f6cb81925a441708521dafe211ca6ca : SUCCESS in 11m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b5f410788c754432a21533dcac7c648c : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3753d152e0fc47048965f49441408466 : SUCCESS in 53m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93fc689e32dd4dcc8ac8d18097494692 : SUCCESS in 1h 10m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08d594ab6ffc470d805f071370d5f7db : SUCCESS in 1h 23m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/220539276d1346a3ad2aad9ee6295077 : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":1},{"id":"788fbc8b70fbb1e4778b3f68678072085832a3c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 05:31:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bf4b8c012be313766264f5abe66f2e55865cae46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 05:32:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677eab8d33ac42c5b486135e822f151e : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"130135dbdff65062059ca60f395c5bb0c8026061","revisions":{"130135dbdff65062059ca60f395c5bb0c8026061":{"kind":"REWORK","_number":1,"created":"2020-02-24 19:23:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/709591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/709591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/709591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/709591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/709591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/709591/1"}}},"commit":{"parents":[{"commit":"7d5bea2164d74471c3cff072f73a45212b3793e9","subject":"Update gitdb to 4.0.1, gitdb2 to 4.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d5bea2164d74471c3cff072f73a45212b3793e9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 19:23:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 19:23:00.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 5.0.0","message":"update constraint for python-openstackclient to new release 5.0.0\n\nChange-Id: I6d1e1316e416fdaf7c94a7d04fb08fc5cd7d6ed7\nmeta:version: 5.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta:release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta:release:Change-Id: I2f9640be40504c9dcd54c6f11f7ac1422240f7f6\nmeta:release:Code-Review+1: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130135dbdff65062059ca60f395c5bb0c8026061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130135dbdff65062059ca60f395c5bb0c8026061"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
