)]}'
{"id":"openstack%2Frequirements~709595","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ic1086fc34770ca20ae6a6f142be2e9e7957334e6","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Ic1086fc34770ca20ae6a6f142be2e9e7957334e6","subject":"update constraint for keystonemiddleware to new release 5.2.2","status":"MERGED","created":"2020-02-24 19:32:36.000000000","updated":"2020-02-25 16:50:23.000000000","submitted":"2020-02-25 16:50:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709595-1582649423822-1d1e633c","meta_rev_id":"c050557f41c50a694fedbf27404515c2cc19c672","_number":709595,"virtual_id_number":709595,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 16:50:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-25 00:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-25 14:47:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-02-25 14:47:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 00:48:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-25 14:47:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-25 16:50:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f334bea7fb1f3e3fb60a56f7cf915c8945108498","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 19:32:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8144cbe6547502e1a1f88e5245d9977ce3220ceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 21:30:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/fac9638db25147d8905d42bda13b3d11 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498bd3b3b11e4ebbb8cdbfdc3c767d8a : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2708e0f6c2b41feaf7035e6cadbec93 : SUCCESS in 7m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d6437eebd98249cf9cef8ed3cc2e340b : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e67b90d6513431383433827975016bb : SUCCESS in 7m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/89da0fa2cb344628a56214ea14a18467 : SUCCESS in 7m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ec474e32d1cb4e30908295dc1806f7fa : SUCCESS in 9m 16s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/486de91a77a54a6b97816f247b2eb6c0 : SUCCESS in 9m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4379dc9e16574eb792010b50ec5d40cb : FAILURE in 8m 40s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5775ca62f9d244f0a8da70f3f0578665 : SUCCESS in 12m 00s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1279c857e58f4e3aaeae06ac1ba0794d : SUCCESS in 10m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b1eba27b913e4d0b8571216da100712d : SUCCESS in 6m 01s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/e15bbe4ee2af4cb292189da404843fe9 : SUCCESS in 8m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c8e7b4d23750416797fe9235755ffb89 : SUCCESS in 11m 37s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/311cb5279c844d7bb5ff6ef47cf240d5 : SUCCESS in 15m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/be93e2f75f1b4026a8426c76d7594ecb : SUCCESS in 20m 50s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e012cb8968164b9b84224e364d1ecd36 : SUCCESS in 19m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b830cdf40f6140ff8263cc206745e48a : SUCCESS in 28m 24s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/11050ada163b498fb3d432fbfc6dc67b : SUCCESS in 29m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/97b3677bc90f44eb8c9dba4242d17797 : SUCCESS in 15m 35s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/22c81c395b264bd5a0b3c31c72bc50d5 : SUCCESS in 18m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31da5a2212844fcf9615d8c3d1ee06a6 : SUCCESS in 21m 59s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/74ad21bacbf44badb4defbc22b548f61 : SUCCESS in 7m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/662cf49a33694dc4853bbd9699a662ca : SUCCESS in 14m 42s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/c3681ee446f24724a948e6271df05e2e : SUCCESS in 9m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/483d47328b1442faa23122d5ce63ad58 : SUCCESS in 6m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/840db24d3d1a4733b86bce1e0ac4e199 : SUCCESS in 1h 19m 50s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ca340848bac74ec8b2395409bc9f9e5d : SUCCESS in 1h 26m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d283140a15f545b9acc94702ff002d47 : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a2a2ce13e4e04366fea04a3b8b8d912df8ff35f7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-25 00:48:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49d830ee81f07b282fee284b85f9bb3ce4f5456c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-25 14:47:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b9c5e3f9a59f7d841d53656b3290294bd6dfb13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:48:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c5ee26fa77e8214d65da2092cbb1e5815275480","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:50:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"866bd28a8b8220f58d3e2f7c04896c5df3875f67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:50:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1a2c70d5ebc458cb479acac13ec8fee : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d978dbaa35a45c890a7ffc9daead413 : SUCCESS in 5m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/10a427f7dda5488d8ab70b2a7ed2d93b : SUCCESS in 8m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/151bf7ea06ef4714bfc322682295bb3b : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/873cfd11b39147729dea33835e40aba4 : SUCCESS in 4m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f2e8fb6dd63b4824a32ee05e09039247 : SUCCESS in 10m 41s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/6af1acc5e53d4e06848e73f0d7f3d753 : SUCCESS in 9m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b0ba572e448b45dca096caec47c1b1ec : SUCCESS in 11m 14s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/70edaf99c80041e49d28bb64f745d30f : SUCCESS in 10m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7f9f639b67dd4b878a17138d35b848ac : SUCCESS in 5m 59s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2e32c06bbc32411e9174e7305e6d38ca : SUCCESS in 7m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ca01fb99729f461ab5066edda864d217 : SUCCESS in 11m 00s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d636c44f067247e99ed431852c6edc58 : SUCCESS in 10m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b445971b320f40ef9b6c33d80cfb8748 : SUCCESS in 20m 20s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/3bd9f3667cf04b739e79553568bb81db : SUCCESS in 19m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f0d648a5defb4850a40a076ffea5ba83 : SUCCESS in 33m 05s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/26556c277b984d3ca143070017e9bd7e : SUCCESS in 26m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b4be5fffd08744038b1a0611b90684a9 : SUCCESS in 19m 47s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/dfe1971ab06640618e68c379d6e7bd9a : SUCCESS in 19m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/488df7e091f647f59617efa5c405f591 : SUCCESS in 23m 47s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/07f26c738a1841198aa7be8af76b373e : SUCCESS in 7m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1a0bfd445c114203a09cedf13c8a21f3 : SUCCESS in 10m 49s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/2be21b7bbaa245b49b92b258029543c7 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d1e08df3d9b48f9914b605c2de8607a : SUCCESS in 5m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9195e1e916da4c64878681d9a6364617 : SUCCESS in 1h 10m 47s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/34131e494c17486c935119528d41b45d : SUCCESS in 1h 47m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/899f0a88df3c46229c8370c2c82fcad4 : SUCCESS in 1h 04m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0c2e17647141a280f8acab2094c050c26b3cbed","revisions":{"c0c2e17647141a280f8acab2094c050c26b3cbed":{"kind":"REWORK","_number":1,"created":"2020-02-24 19:32:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/709595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/709595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/709595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/709595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/709595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/709595/1"}}},"commit":{"parents":[{"commit":"a05ebb2c184defdc2fd31edb4dbbfb9b375356f8","subject":"Merge \"update constraint for horizon to new release 14.1.0\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a05ebb2c184defdc2fd31edb4dbbfb9b375356f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 19:32:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 19:32:34.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 5.2.2","message":"update constraint for keystonemiddleware to new release 5.2.2\n\nChange-Id: Ic1086fc34770ca20ae6a6f142be2e9e7957334e6\nmeta:version: 5.2.2\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Kristi Nikolla \u003ckristi@nikolla.me\u003e\nmeta:release:Commit: Kristi Nikolla \u003ckristi@nikolla.me\u003e\nmeta:release:Change-Id: I213a387ec5d262d85790425c1a1b0f5da4134d10\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0c2e17647141a280f8acab2094c050c26b3cbed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0c2e17647141a280f8acab2094c050c26b3cbed"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
