)]}'
{"id":"openstack%2Frequirements~709632","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I8b8bb12001b6aefc624ecf42690352179287fd05","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I8b8bb12001b6aefc624ecf42690352179287fd05","subject":"update constraint for keystoneauth1 to new release 3.17.2","status":"MERGED","created":"2020-02-24 23:21:49.000000000","updated":"2020-02-25 16:50:25.000000000","submitted":"2020-02-25 16:50:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709632-1582649425918-6db76d6a","meta_rev_id":"a50ceb157d19f39d7e3d5a91d8d45413a12c7c38","_number":709632,"virtual_id_number":709632,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 16:50:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-25 06:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-25 14:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-02-25 14:48:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 06:17:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-25 14:48:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-25 16:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e86e71cb6ac991c5b65a0a7cfe707d029ab1c48","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 23:21:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ea99af1affa6ac03c7199ca3d50e835cb5fc225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 01:39:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/32d8cdee0bf84e708e103c9377c5ed66 : FAILURE in 4m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a468de2c6026492688b47400a3f6b332 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d73051714344cfdbed98ce8e1585813 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91d731e8042343a883f97c0e8905764a : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50472747d9d5484597edd42397563b4e : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d01db989abcf4057836c85295a414e5a : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a5dcba7d74248b5a13bc0f6bea5ec31 : SUCCESS in 6m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b989d63477e9445cb8e912681b89afef : SUCCESS in 11m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8c521e12e3e140c9861014dd7f7f7370 : SUCCESS in 12m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c34f120b064b4b49af3fa08cba2d5d95 : SUCCESS in 11m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e0a0586073d4bc8adfdacc8d2d0284b : SUCCESS in 10m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e1cd9a8f0af14453a6ac752f33b75ce2 : SUCCESS in 8m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4363a844dd3e46b0aa307f85b732ea7a : SUCCESS in 8m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c6d32af4325f4808a4cfaa5119f315d9 : SUCCESS in 12m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6fd2b15ff5aa4ad19049c0e59463577f : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9396dd1b8987479791bbe70fd8abd89b : SUCCESS in 8m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4753cf0657ed4946add2d952729345c8 : SUCCESS in 19m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/322ef33798084d6f8e51da7af1881652 : SUCCESS in 13m 58s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/67636cdde90849a5a7360a71bff01e5c : SUCCESS in 19m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b33d0257966847c98c8357a855501b73 : SUCCESS in 22m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8e595a36643340669ef3c6a8a0206ad7 : SUCCESS in 14m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/acb130053d444396aea97507612f7d92 : SUCCESS in 14m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/42454d91283146a794e78d93d1691ce0 : SUCCESS in 15m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33aaccba5861478494fd497baaad9916 : SUCCESS in 19m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f841254befe541f78d55b92118b86d5a : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2201aef6d66145cabe8fdb54eb66c275 : SUCCESS in 13m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8f5cd56c81fb4af698b7006da4d1fc8f : SUCCESS in 15m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/852fbb1638254d3091efbce47ca96de0 : FAILURE in 1h 19m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/94507a65f8f04c57b42e50fd3673fede : SUCCESS in 1h 34m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/82329dedc1794564ab6db95b8965102e : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":1},{"id":"aba1d01b50b1a253007aac8143debdc2f9ca097d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-25 04:22:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"445e716ab51b59335a2f91170cda517355c5c685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 06:03:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/60bdb207a8e4450ba9b7db523e4753f7 : FAILURE in 3m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa4aea208c84636ad3d8cb53560f41a : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b167540caa9c41d384e20b1540fca117 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/875751c6e9e04cc6bad16ddaa51f3368 : SUCCESS in 13m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea66abbfcfaa4f5e9ae90ed92eb3917c : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39d28b974eee48cf8af550ff3967d02d : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a19b29d144b4c9e882a67076ed5f512 : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/270e3fea1b1047c9a989c5aedcc4c848 : SUCCESS in 7m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c69e8f4eed0f450682919535caa49cfd : SUCCESS in 7m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2822707b140e4ffdbd45cc8531aa0b79 : SUCCESS in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7f984e1a011b4742938940ffb58458c0 : SUCCESS in 8m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/89cfc8ac448f428a9f1fbef6f2087040 : SUCCESS in 5m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a74752df03804317bb6c266307394d4c : SUCCESS in 5m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/86d08730ccce458cafaeed7f0f418833 : SUCCESS in 18m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a0540a39f4d646e9bffb9a9e471a5512 : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bb7e8e0d2c94a6da6060054f23554fe : SUCCESS in 7m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/98f4bec9f8874dd4b77cd5d037c0c93e : SUCCESS in 13m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/059e0f38c51e474e9b0c1081373807f3 : SUCCESS in 12m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f536c3c419ea40189dc21283c8f66344 : SUCCESS in 26m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/be7a266a6ec74105b91009c7eab1c621 : SUCCESS in 23m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1ec6ff0521f9490798b10c03779804de : SUCCESS in 12m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cf854024623c4a49bcfe231fd6a528a4 : SUCCESS in 12m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3c9276a3d55446adb1a48eef111a1f73 : SUCCESS in 14m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df2c222447534b0aba1ea2ca346498d3 : SUCCESS in 21m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2bc9d18111a8404cacc41e415e1c7cc4 : SUCCESS in 5m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e0c08a32c4754278a016ef1d2eda7792 : SUCCESS in 10m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a3dd152094124e638c2d9facd2d081de : SUCCESS in 9m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/edcb49f474f24a66817199ab0845f896 : SUCCESS in 1h 09m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/872c61bd533b4bd5809223286ebe37d2 : SUCCESS in 1h 35m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/565ee80983334294a26e0498c6c586e3 : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2beab5e81f7663433644f2ea96a551e13bf2c553","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-25 06:17:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5029dfe335eb04fbabfd1d13bbb4154bdf3ea412","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-25 14:48:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"420dc1d6abb1f4b35d456a4bd90ad051aabdc76f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:48:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1783055dd3a4ca401847a98c422512925df1f039","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:50:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3b060ecbf2165639a07b18bef3deb5424f91cf40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:50:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d4926c317b4097bab77ffcfdcdbdca : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/233b2067e97e41acadf4bc5b6edd5ed7 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fbf81b6f8304ab9b2988e74a7320467 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/332366d3bf8841ee80f496a75524be65 : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ef3395217324faa8f74dc7d34c3067d : SUCCESS in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3583e9a947b460e8fa98725088f04ac : SUCCESS in 5m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/20018c4704384891bb42da5fc6e605df : SUCCESS in 10m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2aafc474207344a0a46e89550932fb77 : SUCCESS in 8m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/417304860dea41d49209b902630030ab : SUCCESS in 12m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5fad04ce9c07460e91331fe127b2668e : SUCCESS in 12m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/800f7118ce8d49a6a38e3e1e3944c6c4 : SUCCESS in 6m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/abcbdc72d69944e7830e325bc773dbe1 : SUCCESS in 8m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/78d3bee869714608b159f7b319922861 : SUCCESS in 12m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4716d88cbc774149b4ffaa106c53a243 : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf678db3953e42309adeb75e7239b3f2 : SUCCESS in 12m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7fbf7a184628425a8a3825c3c53f3db2 : SUCCESS in 22m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7878695034364a0ea55e414f6c22f349 : SUCCESS in 20m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/821feb67782e435caf4ce2a632543c07 : SUCCESS in 32m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a7c3382453444695b37741b070ec1c70 : SUCCESS in 28m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5991a0cdac534a0ab4a3eb4923009472 : SUCCESS in 16m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/43976f3b25ab4af7ade0fd255635106e : SUCCESS in 15m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d1ca51cd99334ef18978ac0dbd0ad5c6 : SUCCESS in 15m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b3cf24622804b22961154fd8de4f110 : SUCCESS in 22m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9278676cde544e4b63fdb837b7a3248 : SUCCESS in 6m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bede4df7054443d288974699510363bb : SUCCESS in 12m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/268f334df78e441aba66f36fd64ebdee : SUCCESS in 13m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bff6deed9c72465d9f8d33b16c541da4 : SUCCESS in 1h 13m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3259744d862a42a7bb2cdaebc858af17 : SUCCESS in 1h 44m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8ccc4f1089584965a8c813b6e0f982ca : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fa235bb42386f704dfdd8e24e264f42fd8be386","revisions":{"3fa235bb42386f704dfdd8e24e264f42fd8be386":{"kind":"REWORK","_number":1,"created":"2020-02-24 23:21:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/709632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/709632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/709632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/709632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/709632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/709632/1"}}},"commit":{"parents":[{"commit":"c669fbb2b7617b8b8016962ea9b6e52b0f156e3e","subject":"Merge \"update constraint for sushy to new release 2.0.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c669fbb2b7617b8b8016962ea9b6e52b0f156e3e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 23:21:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 23:21:45.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 3.17.2","message":"update constraint for keystoneauth1 to new release 3.17.2\n\nChange-Id: I8b8bb12001b6aefc624ecf42690352179287fd05\nmeta:version: 3.17.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Javier Pena \u003cjpena@redhat.com\u003e\nmeta:release:Commit: Javier Pena \u003cjpena@redhat.com\u003e\nmeta:release:Change-Id: I6eede4c8a29f7accaf8bb764e4454823258232d8\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fa235bb42386f704dfdd8e24e264f42fd8be386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fa235bb42386f704dfdd8e24e264f42fd8be386"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
