)]}'
{"id":"openstack%2Frequirements~709731","triplet_id":"openstack%2Frequirements~master~I7597c3dcc6e1b39060b09ed01a93ed93f757dc27","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7597c3dcc6e1b39060b09ed01a93ed93f757dc27","subject":"update constraint for oslo.log to new release 4.0.1","status":"MERGED","created":"2020-02-25 14:23:58.000000000","updated":"2020-02-26 22:31:02.000000000","submitted":"2020-02-26 22:29:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709731-1582756143171-b95df229","meta_rev_id":"bc83ba92acce08e192bd0b073e30ee110ae48f71","_number":709731,"virtual_id_number":709731,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 22:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-26 20:41:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-26 20:41:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 20:41:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-26 22:29:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32ccbfd6b6cf05de3e2e12d8fd0510f40032970b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-25 14:23:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1630121ff480b180d2a96b031f2d5bb9540cecf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:04:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84cb456856f5459abad1b3899a2bd13f : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af603fcbffc44fd8f728304e1af4abc : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/098d704fb5e946e693572af840ac7d6b : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fa443e489c74a67b11a1c8d52302a9f : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfebc84aae9148678468bec085267923 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6993ccebe5f74d3daf010b0588c64eaa : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83a0f2d2729b45e6a5686a84407500de : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/396423fec619427db72603bcaa7fbc3e : SUCCESS in 15m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1edaeac8c7ad440396fabec08e7c8fe3 : SUCCESS in 9m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/385518d9ed4d475ea17543ab4249b27b : SUCCESS in 7m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3bc875982d4b42e9b7a745f7e1752ce6 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce7d98b52089490694dca24af9773b9a : SUCCESS in 9m 03s (non-voting)\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/47ad008d5f2c41af9eea123c97b67abe : SUCCESS in 17m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a4ba02a79b1448d29179ad0eaaaf86d7 : SUCCESS in 29m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1a60bf4e2e2c4665b323fa7ddb768e5f : SUCCESS in 13m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/032ad0f9b1b44c859f9939f3862c21cf : SUCCESS in 16m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67e7152e78da415d8c02e9b0cb5c0d05 : SUCCESS in 20m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/186d8b7eab494e6984b2e3ffbaecbf4d : SUCCESS in 6m 56s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/921c0d2dff2b45ebb341a65e68753537 : SUCCESS in 8m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38edb8cdf98c40d0bd1264056726eca4 : SUCCESS in 10m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d65813c202ba4854b1c698278312f269 : FAILURE in 5m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d24a506c52bf4f73bc13424fab43d967 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e17cfb9721fb4c1db85dad453810c5e0 : SUCCESS in 56m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b706c61e7374c79b28fbfd43205c522 : SUCCESS in 1h 02m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff6bfb378ad94a50a59482e29140ebb2 : SUCCESS in 1h 37m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/98005ce4aea94517be730d67b99ef30f : SUCCESS in 1h 06m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8fe55f78a99225e8deb391f4896228eaa68b186c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 16:09:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5309fe444f7656164e5734684168d12fc5838eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:59:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e06d9b4b1e49f98463583eb12957ce : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/136fb486cf4b4083ade2f051f112f50c : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/701b70dcefd74536a3bacdae09f45584 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70be4a5137c64ea3b2593e53ce73fc42 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54106c5da11a4792bb425ae5f388b385 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5197471e908401abec7c4e567aaf26d : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c225f668f5d04c97bc5df852cde8a296 : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b595f8a27bf943d88404c7a5b7b08af9 : SUCCESS in 8m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/01158e558ae840afa947487817b9c857 : SUCCESS in 12m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/50f79c6cf04b4777b7edca7bbe7fc4b5 : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f81289c2db2542a5b5719eb6846b8357 : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/156f747dc4dd44018b63d58ec20d7b3f : SUCCESS in 8m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/85f698d82f864feba0e2e09c7915b976 : SUCCESS in 15m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7b6ad73550e4d83b2c1c53220fe4101 : SUCCESS in 27m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d9dfec72fef14c89a1e81f44705db3ab : SUCCESS in 14m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b572f2050e644a79a7e292c473ecf64c : SUCCESS in 15m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51cba2f9ec3a4c209fe8835d7fae0b25 : SUCCESS in 18m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5bfe3452b34847eeafd3241c4849c4b3 : SUCCESS in 6m 10s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/46d5a89b908b4b328fbe97a5eee6be0a : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f515e0e8937946c390c23d9e7a0c3fd0 : SUCCESS in 8m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/96bafd41b2c94bb0aad9ca97e6a0ab9c : SUCCESS in 11m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d358c779127a47c69245434a5ad69ea4 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34d03b0bf1c34421b3cf1d035cd41ed1 : SUCCESS in 52m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6c51d56e2934b03802335b6d6c46b1d : SUCCESS in 1h 18m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/631e76401c3b402a9ab0502e46a7702e : SUCCESS in 1h 18m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5da298a7372144459dedbd3469d9d56f : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":2},{"id":"6939addb2814fabbd3ef6d240cdef7cd2b2952b2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 20:41:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3e7d1abf04367ef87fc76ca422e3852a865beede","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 20:41:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61747c69421a7e2b3e8131c5ca79e357d634bb06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 22:29:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55930aa379bc48b7a36617e41550d338 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abf441ff5dbc472eaf50601255c61608 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cefb61c672334f2e8bb507100ec93752 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05d978014b4f4faf895be3579080fa9f : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/649c702d807f45eab7882d97d0f2df85 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/394eb621948a422b8836db9dd6d9cb86 : SUCCESS in 5m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/43ebc873217d4acaaaa1977d6216ac41 : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4b7a4006d13c410baa18dd41d5a89961 : SUCCESS in 11m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/22813969b65a4e34b2b81297e790f37d : SUCCESS in 8m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/691e23e7224d42baa83d1e272be26d3c : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f16b545e882e43da82b006f12d32f7a2 : SUCCESS in 8m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/968a6d1f940542718bdb914de8f18cd6 : SUCCESS in 19m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/452855ce5bfc47ed9753c698bb196cee : SUCCESS in 23m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/631a15c95d6843b0ad8229a29bd40b70 : SUCCESS in 13m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1e1e784ce4e64f24b8365540ba76eb76 : SUCCESS in 15m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63755b2f8d7a4fbfb2935f5e9a9326ad : SUCCESS in 16m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/db17143d72954857b41f8bc9cd44e3b4 : SUCCESS in 6m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/473d3376bd5748cbb56789d924136148 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/97f9b064172a4d36a91b70af77d47159 : SUCCESS in 7m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6054f049dbff46d6a149d22613993633 : SUCCESS in 11m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b048448481d449b58c1fca3e3fb14aaf : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e40c75b080b468289ba442cea193272 : SUCCESS in 1h 08m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1042c94c143a46d9a599123f4f265625 : SUCCESS in 1h 08m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93b5a327561447679b7e6b3f332f27f5 : SUCCESS in 1h 17m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/06bccbbc26804d88a5e4addf4398c008 : SUCCESS in 1h 06m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4d0ebdba992deb37da87e26d2c1cf07524a348a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 22:29:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"30824109030483172940eb3898ba22273563471f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 22:31:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e09546753d4e5199988286a9adb9fc : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06899679c3c5b8861ecc30d54ef8a1d7a71f5b67","revisions":{"98b29dd50ef115573e14776b6b67bc1ff688358f":{"kind":"REWORK","_number":1,"created":"2020-02-25 14:23:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/709731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/709731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/709731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/709731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/709731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/709731/1"}}},"commit":{"parents":[{"commit":"130135dbdff65062059ca60f395c5bb0c8026061","subject":"update constraint for python-openstackclient to new release 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130135dbdff65062059ca60f395c5bb0c8026061"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-25 14:23:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-25 14:23:55.000000000","tz":0},"subject":"update constraint for oslo.log to new release 4.0.1","message":"update constraint for oslo.log to new release 4.0.1\n\nChange-Id: I7597c3dcc6e1b39060b09ed01a93ed93f757dc27\nmeta:version: 4.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ib1ac18923476f4206b2b4531139476b7b21a9d8b\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98b29dd50ef115573e14776b6b67bc1ff688358f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98b29dd50ef115573e14776b6b67bc1ff688358f"}]},"branch":"refs/heads/master"},"06899679c3c5b8861ecc30d54ef8a1d7a71f5b67":{"kind":"REWORK","_number":2,"created":"2020-02-26 16:09:04.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/31/709731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/709731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/709731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/709731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/709731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/709731/2"}}},"commit":{"parents":[{"commit":"130135dbdff65062059ca60f395c5bb0c8026061","subject":"update constraint for python-openstackclient to new release 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130135dbdff65062059ca60f395c5bb0c8026061"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-25 14:23:55.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 16:09:00.000000000","tz":-360},"subject":"update constraint for oslo.log to new release 4.0.1","message":"update constraint for oslo.log to new release 4.0.1\n\nChange-Id: I7597c3dcc6e1b39060b09ed01a93ed93f757dc27\nmeta:version: 4.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ib1ac18923476f4206b2b4531139476b7b21a9d8b\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06899679c3c5b8861ecc30d54ef8a1d7a71f5b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06899679c3c5b8861ecc30d54ef8a1d7a71f5b67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
