)]}'
{"id":"openstack%2Frequirements~709740","triplet_id":"openstack%2Frequirements~master~I5fce18a91fda7a62c2b09bc9f6a588160b8f7e81","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5fce18a91fda7a62c2b09bc9f6a588160b8f7e81","subject":"update constraint for oslo.middleware to new release 4.0.1","status":"MERGED","created":"2020-02-25 14:32:49.000000000","updated":"2020-02-27 05:57:33.000000000","submitted":"2020-02-27 05:56:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709740-1582782970090-dc82b96e","meta_rev_id":"bc2ab30315ecb05912c8a90bf3b85d3e6d6d62c7","_number":709740,"virtual_id_number":709740,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 05:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-26 22:25:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-26 22:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-26 22:25:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 22:24:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-26 22:25:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-27 05:56:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"608d6af44af00258aa63d562f8610683c7dd0388","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-25 14:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c613ef4572365282faf69b1efa796b97726a4ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:12:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4427111bb8904d139dd99715201c87b6 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a14eb6040c48ceb9d6f311205cb81b : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a194fe8412ce479c99412234cb570471 : SUCCESS in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4310379b02c047fe86f409c6b65efbc6 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a7d9538272048d6bd836939c87e12bf : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/036f8fa3524a4633994fc971756c467b : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6e62b11aec54f6bb75cb2e1b3710f34 : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bc0d3f0bcb8e4c64b433c81b14a0e40a : SUCCESS in 8m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1495ef8fd16544f4829feeb8d3a3502b : SUCCESS in 14m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/03a7588409ef4c74b0ab219234553bea : SUCCESS in 6m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e98c0ea150ea4b09bc3a1ce95d68b129 : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca71515f18224e4fbe49dc21417d1b77 : SUCCESS in 9m 27s (non-voting)\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e713b3cfbade4468835ed27c313b3121 : SUCCESS in 18m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aa41968652f1431590675fdc6d33abc9 : SUCCESS in 38m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e5f12a61d2694265b57ec94f3b6dbd85 : SUCCESS in 13m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d37e5292bddf424cb9537e8e30e17b47 : SUCCESS in 13m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a0eda8a301e4c0a803249fc2cc9636c : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2d00fe152b3b4592bfd725c025a5d281 : SUCCESS in 5m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/691ce7f02395434f81f9e648b4e81a41 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e6b3aba8fad049a69aa022eb98f5bc38 : SUCCESS in 7m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ccc52946716e4623b56286f4ccc3adb9 : SUCCESS in 16m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3206ccb9902040e0b7b8460cf370497d : SUCCESS in 13m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/554c091974a4499eb86d2440de65c1d6 : SUCCESS in 1h 07m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a51936486e97428281f50f464a05eddb : SUCCESS in 1h 14m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95022da6460e402a88ec9652037dea9d : SUCCESS in 1h 25m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c8e2d1539c7d4d85930c6d9d6980dd81 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":1},{"id":"712029f0fea34f0f807beb21c4db2f22e023b91b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-25 16:27:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c64493032c535c51f02c2e7383720f11006c082","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:28:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96f60cf29b62b65031109cf809ec3855b42315c0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-25 16:28:27.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"91060c49e5725088e768d35e14a53cac1b81005a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 22:08:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/396e1738ff8142499bd20974cce87cd6 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ad75069c3a47aeba07e983873a19a3 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9992478664a542c68d4352115aabcca2 : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33968d7c384441ecb8402f031510f1ad : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cd95f48ad0846d9b8365291a7a8e18f : SUCCESS in 6m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79b9cd7172c04f67b811d7f1d32d3610 : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3f7d8c3c96c04fc3889f69b87d09b13d : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/08a9a1a8c1084747bb06645eccc75669 : SUCCESS in 12m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d7a3a7e6e577407288a2b450ea246ee9 : SUCCESS in 7m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5a477cbc413245178619f0af17879afe : SUCCESS in 10m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cd8e9bcdc801462090e9eee4c92b478e : SUCCESS in 22m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7985ad1490ff4072b840f75032862e2b : SUCCESS in 26m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/03bfc4e7623f4c048275d8c3aec52289 : SUCCESS in 14m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/536d68a67b834f6a9d1f21db89c3d184 : SUCCESS in 18m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7bec39b81be4b4a9728a179fe05ea04 : SUCCESS in 15m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a80b43ff37a0428eb8c01a0d0a002e27 : SUCCESS in 6m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b4a9a91459f04fa587277734fab75c67 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a14ad79dce804d33a5c6c22abe7d9e00 : SUCCESS in 10m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/553eaaf260b8449eb9f7d7bffac28730 : SUCCESS in 14m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/21de2bf4eff543e7a548fa6c6a86aa99 : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd13e96cb96d4170acd8e41eed657ad2 : SUCCESS in 1h 04m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/318ed68f7b5d4b91a2606458a8d0a498 : SUCCESS in 1h 19m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1cc5aa13ac0d4e92ba10fba69dd315c1 : SUCCESS in 1h 19m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/37bb1c9073e1481686e4774cb259912c : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5767d058e53cacd8b02214fb6323fe4d9c19e8f4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 16:00:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e4a766738205708182dd332b56e4c200506429b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:35:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2092c7d86540688fdd2aa8b3fd4ac3 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d458b59e8a4f438817ca64a9af39a6 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4721615c3474ffba340002f86f6c860 : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c926592f0c194fc89333991f7147cfbe : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b99cd04f34b48b7b4e7bece7b3dd4e9 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3851dd944e6f4ec7ac5e06a19b08dd15 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75b60b8d57ea433a8f78613e7817a4e4 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3b7836c1df2f4d95ab5d0fd864da8328 : SUCCESS in 14m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/508561ddb5584dc4bcc606f5a584ae12 : SUCCESS in 11m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e351875712ec4830ac04b0a94344a8b6 : SUCCESS in 6m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d58e44c4336a432c929511f487529277 : SUCCESS in 13m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27cf3a026a184a98a4651eb0426e451b : SUCCESS in 8m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1e8423415b9e4c518c0b7360ebe8e8c6 : SUCCESS in 15m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b290bf3a70a344789ae4ef8b6c83e8f7 : SUCCESS in 27m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f494b9cbe0bf413898ce379325680b94 : SUCCESS in 13m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f8431b087828417a8e9486bc25ca95db : SUCCESS in 15m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70609128711745a485c1206cbcc56d17 : FAILURE in 19m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ebd652ad239c49faa91e91ef7b51d61a : SUCCESS in 5m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9b1fc6ed444343099f134c523af4b16f : SUCCESS in 7m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8cd98e5a8e0f41609561c316e085c517 : SUCCESS in 10m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c0ad2bdfdeba45ed99f84ed4744a9f9a : SUCCESS in 13m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8cea0a28ccf7496093e2a9d661aae7ab : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03e2e41424e7439286fb3b87b9c44b47 : SUCCESS in 1h 00m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/33bd9a9b8cf94bfb87a204c4de4612b4 : SUCCESS in 1h 19m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd6e9bfce75a4965982511fb5491c8ee : SUCCESS in 1h 20m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5c2e6f4b79b948c5b30c58945054bb68 : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a7b9a2b63f5d67012e81f8d67401db80e6054291","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 19:36:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"027f19b559c63533f95f71c535734ad8706cb5e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 22:23:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4bd74fbb3b44f36a7f7fe224a5e820e : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e36539688de4f83aaff21a555fcd63e : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e93cf490c5e43e99cedf072d5313aab : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c32a02f5db6a4bbaa999f2b94a27a5a4 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fa907d1cbd640ca86a510cbda0e2288 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76fc8b5912dc42bea4362fc1f4aa6913 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2738e920ec8c4ce19b5d4adcc44b5249 : SUCCESS in 4m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/73937f956b9a4ffeb3ef4967c76b0034 : SUCCESS in 8m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8ee53a83a0bc495c89cb46770f690473 : SUCCESS in 10m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fd8f6dbae27c409fb59fac6abf359944 : SUCCESS in 8m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/73943e8342174576a8abd60da24bc5f4 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78769c5a928f47f18b5b5e1bcf9d315d : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/41be81ac34644beb90c71bcdeaaf2fd4 : SUCCESS in 26m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e318bc1e461746afade57718fd907216 : SUCCESS in 26m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/85b3009af72f4949b827812a1f9063b3 : SUCCESS in 14m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0006d79f8bdd4082a8bcdd1c7abf6b56 : SUCCESS in 15m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6fb6b47436034cc89f4fd6f366bde82f : SUCCESS in 19m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7f351b1f518748a58d1aaef6fdafbe7b : SUCCESS in 6m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cdb46e872d47490eb3b58688b76a7f73 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d6bae321cda144429910a349fd9403e7 : SUCCESS in 7m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/28145cc6a92c49c7a0c9ba25b83782df : SUCCESS in 12m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4c56368c23de4b17be407aa91e61e338 : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59b57f56b50343899ffb02cc45a2fcb6 : SUCCESS in 56m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a95aaf45d11f4f298fa4fa3d6241597d : SUCCESS in 1h 10m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4545d428b8948bca9fb041e078f38e0 : SUCCESS in 1h 19m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e3f54e56a6744042b543c2fe4589026e : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":2},{"id":"23ad0e0f799e7961240849a6b7da71c98da16a1f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 22:24:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e8de0c174de5df45ffacdb6b4f5e57cf0d13a457","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 22:25:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2d64ba49157e26682f3fd1a19c66bbbf84539421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 22:25:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"24a602f2787ea0f3059f382dd46c8c8f0219aea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 05:56:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a90a77cac147c49b356364967b3eb5 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abdf433282b4157bf543ae149b843dc : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ed163b5964648fd80a8a99b7854f1ed : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbcdb84eab7246e3a17a3ef7d35c71c6 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/138ce9ac3b054360b0caef39e4314644 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e6aa159f5ec4fa19e800093aedc9e42 : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/57e997bda7c34d1fb09d106112e21807 : SUCCESS in 8m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/37852fe5798144dcb9b9b0986ef7ffda : SUCCESS in 11m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4e07f4952061473989dcb1a28182f272 : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5619a78e4a3c4ad18a0825d0864bba71 : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d3362ed505841b0a52be13b8e1230ee : SUCCESS in 7m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/86e08d2f042f435c9266c41d0a645257 : SUCCESS in 13m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02679731632544f29eaf03ceba29af75 : SUCCESS in 26m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f8f15fb38d949358cb3b098f6857ff4 : SUCCESS in 13m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a0d39d9bb8e0487ba6b12b8197c145f9 : SUCCESS in 15m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22e84aa13eb94604869966f0ac5e1214 : SUCCESS in 16m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/326b174a2dfd44afb5c259ce16763142 : SUCCESS in 4m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5c5b27d5a6d6402c9258e7aa47787265 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/276d5dec572e4526acf42265611bb5e8 : SUCCESS in 8m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c935a478ef514eb39f6a92b4b224b241 : SUCCESS in 10m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b2441083e56b46bfa1c9359e50363d13 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/683ba2bde2634cfd9f39b38d01431f55 : SUCCESS in 53m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8dd938b595594541abf0633bcc2b82d1 : SUCCESS in 1h 09m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3fda14d833e49489de9f0eb979f5c59 : SUCCESS in 1h 23m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0806b2242af44b5e9f12e64924b07759 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6e106978c997febf1fb39a0eb7058a61408c6ab0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 05:56:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6f298c01888f56f8af9495077ab47bc7b15d54c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 05:57:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15458edea43d4f9ebc1a8c484a9b6d25 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"538517ea1fadb0e1403fc8beca3b37b2977a3f57","revisions":{"4644a27fed2debad3af4eef669865c5b471134b6":{"kind":"REWORK","_number":1,"created":"2020-02-25 14:32:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/709740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/709740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/709740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/709740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/709740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/709740/1"}}},"commit":{"parents":[{"commit":"130135dbdff65062059ca60f395c5bb0c8026061","subject":"update constraint for python-openstackclient to new release 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130135dbdff65062059ca60f395c5bb0c8026061"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-25 14:32:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-25 14:32:47.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 4.0.1","message":"update constraint for oslo.middleware to new release 4.0.1\n\nChange-Id: I5fce18a91fda7a62c2b09bc9f6a588160b8f7e81\nmeta:version: 4.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ib1ac18923476f4206b2b4531139476b7b21a9d8b\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4644a27fed2debad3af4eef669865c5b471134b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4644a27fed2debad3af4eef669865c5b471134b6"}]},"branch":"refs/heads/master"},"538517ea1fadb0e1403fc8beca3b37b2977a3f57":{"kind":"REWORK","_number":2,"created":"2020-02-26 16:00:42.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/40/709740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/709740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/709740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/709740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/709740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/709740/2"}}},"commit":{"parents":[{"commit":"130135dbdff65062059ca60f395c5bb0c8026061","subject":"update constraint for python-openstackclient to new release 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130135dbdff65062059ca60f395c5bb0c8026061"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-25 14:32:47.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 16:00:38.000000000","tz":-360},"subject":"update constraint for oslo.middleware to new release 4.0.1","message":"update constraint for oslo.middleware to new release 4.0.1\n\nChange-Id: I5fce18a91fda7a62c2b09bc9f6a588160b8f7e81\nmeta:version: 4.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ib1ac18923476f4206b2b4531139476b7b21a9d8b\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/538517ea1fadb0e1403fc8beca3b37b2977a3f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/538517ea1fadb0e1403fc8beca3b37b2977a3f57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
