)]}'
{"id":"openstack%2Frequirements~710138","triplet_id":"openstack%2Frequirements~master~I398c0aee001f4e6a8e7fc89597738b356b885437","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I398c0aee001f4e6a8e7fc89597738b356b885437","subject":"Manual run of generate-constraints","status":"MERGED","created":"2020-02-26 20:30:32.000000000","updated":"2020-02-27 08:20:28.000000000","submitted":"2020-02-27 08:18:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710138-1582791485545-dc82b96e","meta_rev_id":"a084756ddfd62719a8cb38b68cc755391110b34d","_number":710138,"virtual_id_number":710138,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 08:18:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 00:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-27 00:11:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 00:11:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-27 08:18:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"524cca13e2919e5b8b67371298b0c85069771719","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 20:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4adfb4ee0ea93bc8c1d79de1077979b747d5f21","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 20:31:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"da12c813cd3785ee655b1a47f1ff8e39e259ab22","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 20:34:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aa786be644390b04eee14d99d83e47a8eebe5e18","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 20:39:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3a666d59de3af2a5d95a26cced5108de3c96323","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 20:40:46.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"33c42a51ce924323bef79e569f9a66d9376e69e4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 20:50:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ec6453e9c3f8001e00b42e6674a406c40b85f7fd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 20:51:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dff7ad03b5187a8528fb10efcf0c478c5211be32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 23:48:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920639c8818e4791a7ca8cfc78d87f29 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab0e8a86acbb4feda211732ff05da46f : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1d83704b8a3414a8b777767c5a894cf : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c45441e109524c5198903141656987c4 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb22eda9dfd048d7a3fe8a51b8103be1 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cfe4ea58ce74ebfb26955da7eb8f2ce : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6fc28418b7e44cb5a7d638668b6e420e : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6507284cfece45c5ab457b58ddf3b821 : SUCCESS in 9m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/21d37a15513f43ceb0a716e5a1c4c4bc : SUCCESS in 11m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ec18ec1658da47b5ac4563acad99716f : SUCCESS in 6m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ab4151ff98d74909a8c9159d54a72070 : SUCCESS in 12m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/57c99bf9784e4edaa6f828b2f9e2879f : SUCCESS in 9m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/817e7d9285844cd688b8e2dfaa479481 : SUCCESS in 14m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f619f4c44ea94df3b19e47a7ec55e188 : SUCCESS in 29m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be085b2eeda24a6b86b85ca18f45c958 : SUCCESS in 14m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2d7b3c6623344b3aa143d17f23489a7e : SUCCESS in 15m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d923ac335f48448bbfef87ec414a2536 : SUCCESS in 17m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6a5c89e58789412b9b794c5b69b78dde : SUCCESS in 6m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c3ea1c5f76564c8696fb92b7a06231e7 : SUCCESS in 8m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a1f97a1bcf841d6b4cb50f718ded5e1 : SUCCESS in 9m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1b993c48810b409c9b6b38cd07cf8e8f : SUCCESS in 13m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/460e3a72489642ac8e12be40798d27a8 : SUCCESS in 13m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fd14cf55acd43f194b335dbdd1069c1 : SUCCESS in 51m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d545b46b0e74294bd7af7287973ffd0 : SUCCESS in 1h 16m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a7601c5c27e4f9da35fc4e05ff0bf87 : SUCCESS in 1h 23m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/567e99fae54842f9bfa0db5e1b180013 : SUCCESS in 2h 08m 02s","accounts_in_message":[],"_revision_number":4},{"id":"98c9601f4ce44467dc17215792c20b53df7ae14e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 00:11:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6d7a1ff73c45c92ad2373c11f40164f4f989e40b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 00:11:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7e92e187c8e9b47efaa0da054ab3a59f5f85cf8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 08:18:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35510964db34e13b8eae09827801ae0 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052101b31c114ccabb5f2f449a84f1a9 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94741e9d4dc940989a4a5d74278c84ea : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50e4ab04a17c4d479e26df0f8412db9b : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed34c2cd5a034b559af0a16e012d7b24 : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd79ed7f7d8e42a2ac1557b15f685335 : SUCCESS in 5m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8d0f925ab2c443fb8de12e5973b42247 : SUCCESS in 10m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a41d62e96b8c4742812fce29f43d80e3 : SUCCESS in 9m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9eec42cd667b445397d39610132ee80e : SUCCESS in 6m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ce355e59ef24437083ab8c21570a9c6d : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb133d46b66547a498a9953c387f3534 : SUCCESS in 8m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9bbbe2ec3f764200b3ecbded399f9d73 : SUCCESS in 21m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/370d1d918d48458fba29242f6585f016 : SUCCESS in 24m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/67271656c60a4c4db701409d9257e536 : SUCCESS in 14m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/59b9fc99683e45f789084d933cc9d395 : SUCCESS in 16m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de58c7de2e3c415fa7a9d7f62d55e1ca : SUCCESS in 17m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7c2685dba26a41f1a115b161b09b3c35 : SUCCESS in 5m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/322ac25ac96e4557bf3c2dda918897e9 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/92bab0357f744739b54d1808156f1182 : SUCCESS in 7m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/69577a9ab384423abe9ef378482b1786 : SUCCESS in 10m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/789c316b9f674df3b1a9247f457ee7eb : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4216f09b8684421292d10f22a866a56f : SUCCESS in 1h 07m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/839b33de89a0400b8646d2a3b7e0f649 : SUCCESS in 1h 10m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6099fd09172944cfaf38145c748cdbd5 : SUCCESS in 1h 28m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2397494ac8d64a59b2a463f06ea0ec05 : SUCCESS in 1h 34m 37s","accounts_in_message":[],"_revision_number":4},{"id":"e23d04ecdb6672479832f6ac781ff8cb6fecd97b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 08:18:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a084756ddfd62719a8cb38b68cc755391110b34d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 08:20:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb01a4ee4534b17aea68734212163c6 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0933fc1b735481e75184c0c81079111b5e0c47dd","revisions":{"55e75717de8f0daa6a394f247f9337bf660b73e5":{"kind":"REWORK","_number":1,"created":"2020-02-26 20:30:32.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/38/710138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/710138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/710138/1"}}},"commit":{"parents":[{"commit":"f7aa34359c03ffecfe74a5377839e7f8904b32cd","subject":"Merge \"Set tox_constraints_file for cross-jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7aa34359c03ffecfe74a5377839e7f8904b32cd"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-26 20:28:55.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-26 20:30:19.000000000","tz":-360},"subject":"Manual run of generate-constraints","message":"Manual run of generate-constraints\n\nhad to run manually to fix the python3 cap on gitdb\n\nChange-Id: I398c0aee001f4e6a8e7fc89597738b356b885437\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55e75717de8f0daa6a394f247f9337bf660b73e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55e75717de8f0daa6a394f247f9337bf660b73e5"}]},"branch":"refs/heads/master"},"1785e00dafed5ccfc8b0cbc0b131d6ea51a4b596":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-26 20:31:10.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/38/710138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/710138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/710138/2"}}},"commit":{"parents":[{"commit":"f7aa34359c03ffecfe74a5377839e7f8904b32cd","subject":"Merge \"Set tox_constraints_file for cross-jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7aa34359c03ffecfe74a5377839e7f8904b32cd"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-26 20:28:55.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-26 20:30:54.000000000","tz":-360},"subject":"Manual run of generate-constraints","message":"Manual run of generate-constraints\n\nhad to run manually to fix the python3 cap on gitdb\n\nhold back docutils because boto likes to cap things\nhold back msgpack because salt likes to cap things\nhold back alembic because it breaks nova tests\n  - review.opendev.org/708707\n\nChange-Id: I398c0aee001f4e6a8e7fc89597738b356b885437\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1785e00dafed5ccfc8b0cbc0b131d6ea51a4b596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1785e00dafed5ccfc8b0cbc0b131d6ea51a4b596"}]},"branch":"refs/heads/master"},"66f244851ca38f3323791187cfec966342dd136e":{"kind":"REWORK","_number":3,"created":"2020-02-26 20:50:43.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/38/710138/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/710138/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/710138/3"}}},"commit":{"parents":[{"commit":"f7aa34359c03ffecfe74a5377839e7f8904b32cd","subject":"Merge \"Set tox_constraints_file for cross-jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7aa34359c03ffecfe74a5377839e7f8904b32cd"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-26 20:28:55.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-26 20:50:35.000000000","tz":-360},"subject":"Manual run of generate-constraints","message":"Manual run of generate-constraints\n\nhad to run manually to fix the python3 cap on gitdb\n\nhold back docutils because boto likes to cap things\nhold back msgpack because salt likes to cap things\nhold back alembic because it breaks nova tests\n  - review.opendev.org/708707\n\nChange-Id: I398c0aee001f4e6a8e7fc89597738b356b885437\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66f244851ca38f3323791187cfec966342dd136e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66f244851ca38f3323791187cfec966342dd136e"}]},"branch":"refs/heads/master"},"0933fc1b735481e75184c0c81079111b5e0c47dd":{"kind":"REWORK","_number":4,"created":"2020-02-26 20:51:58.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/38/710138/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/710138/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/710138/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/710138/4"}}},"commit":{"parents":[{"commit":"f7aa34359c03ffecfe74a5377839e7f8904b32cd","subject":"Merge \"Set tox_constraints_file for cross-jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7aa34359c03ffecfe74a5377839e7f8904b32cd"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-26 20:28:55.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-26 20:51:53.000000000","tz":-360},"subject":"Manual run of generate-constraints","message":"Manual run of generate-constraints\n\nhad to run manually to fix the python3 cap on gitdb\n\nhold back docutils because boto likes to cap things\nhold back msgpack because salt likes to cap things\nhold back alembic because it breaks nova tests\n  - review.opendev.org/708707\n\nChange-Id: I398c0aee001f4e6a8e7fc89597738b356b885437\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0933fc1b735481e75184c0c81079111b5e0c47dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0933fc1b735481e75184c0c81079111b5e0c47dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
