)]}'
{"id":"openstack%2Frequirements~710204","triplet_id":"openstack%2Frequirements~stable%2Frocky~I419edb985683c3f5c96d98833bf39dc523837e90","project":"openstack/requirements","branch":"stable/rocky","topic":"tenacity","hashtags":[],"change_id":"I419edb985683c3f5c96d98833bf39dc523837e90","subject":"Bump tenacity to 5.0.1 in UC","status":"ABANDONED","created":"2020-02-27 09:04:02.000000000","updated":"2020-03-05 21:34:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1fba47c5a3e59bd21b37ec58dcd485132d1a0644","_number":710204,"virtual_id_number":710204,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-04 14:05:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-03-04 10:30:16.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2020-03-04 20:46:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-02-27 09:05:31.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 09:05:31.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-03-04 14:05:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-04 20:46:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9d474cdb2b68e17b27243d1b7115c1678ed70d91","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-27 09:04:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f722be7202494bf96999cb2366776b4ad22f9af2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 11:49:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/94586946a49f4aeba8bd1272e5037c84 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c629a7b06847aab7a3df8ab15409a6 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/251ce1ab2501474a9fc6fbde7f5d772f : SUCCESS in 5m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db1db31af6704e90a9a57ea139de596a : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/604b1094577c4b77a457fba8245dfd28 : SUCCESS in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fbb01867713644db9dde4c362d362e89 : SUCCESS in 6m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/45924759fab448938afc906a8239562d : SUCCESS in 12m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/878d53a9f7f742659812af02ee212431 : SUCCESS in 11m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7affcc5e29624586b11220b106282177 : FAILURE in 10m 33s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8a7c8b0339f94c57bb25bce110f0240a : SUCCESS in 11m 24s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a88bf778f109468fbc797d9c74d7679c : SUCCESS in 11m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b76c584c36cc4f64bccce0932154ad8f : SUCCESS in 9m 39s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/165a93d8cba84d37b1367c69870f62fd : SUCCESS in 8m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/de5fdf5313834c66ab0facc3f2061646 : SUCCESS in 10m 55s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/0108984f0d454afeac925d00f4ca3954 : SUCCESS in 14m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/06bf30a3f2ed4e28aac2039463db31f0 : SUCCESS in 15m 54s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f2a22e58d33e408880dcde9ad01c6b70 : SUCCESS in 17m 58s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/42762d2ffde342c697779ffcb53c3ac0 : SUCCESS in 23m 09s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/37b219567a7f4ddb92e1f10c9e394833 : SUCCESS in 24m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bd14a289f12a4127b044e59f1a73fe32 : SUCCESS in 17m 08s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3b20ac9626d247a89c95ce3a8ba73e2f : SUCCESS in 15m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be34de30b6874d629a8ecb62e33572cf : SUCCESS in 16m 59s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/ab70e0d08d5a4894ba746f567868a43e : SUCCESS in 8m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/92fdb8020b3e4e86815a88aeda94e031 : SUCCESS in 16m 50s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/3328499a611d49cba29a582e76de67bf : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a595ca9279d4a02b7bff5ab7c69a161 : SUCCESS in 8m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b1a2e2bb9c94589a526b7502ff4ffe0 : SUCCESS in 1h 32m 34s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4b183cd5b6854a5ebb7217527f3fb208 : FAILURE in 1h 33m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e4beacc0ba5e45c791c8bd857579031f : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cb55e07340205a85e315d37b9bd82f45fa32ece4","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-27 11:50:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"699718e8e932023346e0fcc650253cd0b60f2316","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 22:05:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b31a5f78cc43c94d4e8f97073c54c51bc596ee70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 22:44:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8a002c2eab034891afa837719d292bb5 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48268ec65d184fbdac230ca1e3faa4dc : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19c264b4b9314ec1b0bea48fd1d87b36 : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4025bf77786740b285829fcb190515f1 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee5f2228f3d5448ab88fc4f97d1578a2 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71e1ec65253543c2aa8e3744ea4ae5e2 : SUCCESS in 5m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8f74f6e1bf434570979fd26a8a6f2130 : SUCCESS in 8m 58s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d9f424516dad42ee90779673285c3339 : SUCCESS in 18m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b0e0162b22414bc4ace0191b2b816081 : FAILURE in 10m 57s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5f823aa275314ecd8eacd6dc44349520 : SUCCESS in 12m 31s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1a404da1e0654811ae5d88cf30ff095a : SUCCESS in 10m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/61af0709e369488a99ed82740244a8b5 : FAILURE in 6m 00s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/60bc6ff09b274eecb86552e470156c2b : SUCCESS in 6m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b111c4d9125441f5b3e789fed5dc9c2e : SUCCESS in 10m 31s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/2b9c78fc1ec4422e83f1a26e0ff1dbba : SUCCESS in 10m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/23c2de94ce16427182c4852a4f478249 : FAILURE in 20m 14s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a513dc3f477f45999598060cfaf2d564 : SUCCESS in 14m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/10d4cd05d5194035bfc14b911611c398 : FAILURE in 27m 20s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/3754b1bf3025487697eb9343a864b7ff : SUCCESS in 28m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3394c66c2698409488805c333d499bd6 : FAILURE in 7m 41s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/85bd10d034d74e34a0546cb6413c408d : SUCCESS in 19m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/885e8d5d9cc2467f9838fd167ae537e9 : FAILURE in 8m 23s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/ddc2bf6440af4fba859db99cfd57bb0a : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5a658de051ce483fa35c20c0689cf2ff : SUCCESS in 11m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/517eff771e58434f856d48d6676d2b20 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e271dfba63d48f0af1a8ef750f951f3 : SUCCESS in 4m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8821ce3331b4298be9e8f8acbfb5462 : SUCCESS in 1h 21m 16s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/fdb4b30b1b1a4a10bc8d30622de11368 : SUCCESS in 1h 35m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/674a8d3596f74d1aa3ef168091c0e44d : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2b62ded5dde68d2d4a50a5e71423b0e1ec258e78","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-01 07:27:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10f08af6844cdc03cd04e9e3ea6d33473272c3f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 09:04:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/324ea06c41c4422685b54020652833a1 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e9d73572264a72813ef38bbb3755b5 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2294286abe64292a5fa85328c0a3176 : SUCCESS in 5m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b9060fb671243c691f76e16955789b7 : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8814f26d2a8495983ddb2500a25a4ac : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9da4b4d7cd5d4840b4ddb51b6cbdd19f : SUCCESS in 4m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ac3f40a1caf04a17b5843957531e9d62 : SUCCESS in 8m 09s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/342fa701e4a94a568fb9ca53165f054a : SUCCESS in 8m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7636786f20824f70bf7e819ef3536a3d : FAILURE in 8m 50s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c5f4fb992ff94ea68473c917da088cc4 : SUCCESS in 11m 51s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/33e42000b3c146e0a665c232d9b91885 : SUCCESS in 11m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/99d34038cb6f4c2d9640086c491df038 : FAILURE in 6m 45s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/1f327c1c80444eee93edce6d85ad2b30 : SUCCESS in 6m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/332d1466b322499cb22135aadcba6612 : SUCCESS in 10m 30s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/082448b4db0f4228b437bed69e3de320 : SUCCESS in 10m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7882c40e4f14449a9f0284fbf71bde84 : FAILURE in 16m 06s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/7f6b6d31247c4f63913759a057c0ff24 : SUCCESS in 13m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a221a24e427e4d62ab1b1e91f76b48dc : FAILURE in 25m 36s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e718c273d82f4b8bb98cb4f0aeab2c7f : SUCCESS in 25m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c4b6c842249349aea1788b8834500e66 : FAILURE in 6m 29s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1e07ba6dc4f3464dbd9cf5d9d08203e5 : SUCCESS in 14m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e08d5129ab414b8cb8fa8cf82cd44f4c : FAILURE in 6m 55s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/ed20e57bc07b4e6d80249fe4f3165ab0 : SUCCESS in 6m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/398814f969804162954041a8c5cb539f : SUCCESS in 9m 06s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/890d62738c144c7598d14318d7d04824 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e734474f8fe4b36bc73c3fcb93043e4 : SUCCESS in 4m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/caf049cce0454119979738e607a91047 : SUCCESS in 1h 08m 31s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7dd232e428c2422aad4b01b961ea021b : SUCCESS in 1h 32m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c5d53a52afb14c2f8a3992706778b1fb : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":1},{"id":"22596f91a5f29252d67ab4863ab8242bc1e68b55","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-02 21:13:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5c1cf9066215d020e338f2b22f0008408fa731b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 00:11:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c1bf8cb505c545aea2e6646da3a809dc : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccde46b79b944426bbe3dfd0aa3bf3b0 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84a9aeffaa2d4faf873757fa7765d27e : SUCCESS in 7m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/73c65dc10ec543238452410c2e6acfba : SUCCESS in 7m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5beb4dac4b714ed4b8e95663fa281fd7 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25cc5ed1d50846e3aec4999392a1f883 : SUCCESS in 7m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c4d9c22971a84028975e8a1c60e1491b : SUCCESS in 12m 39s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f40e6e64792d468fb0278ef226fac0e3 : SUCCESS in 13m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d3da7e859a6f4b4791b0f963a2fc3009 : FAILURE in 12m 02s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/430b38a81e864d319575cdd796528d02 : SUCCESS in 13m 11s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/e8798baf0f734f559e9bd9aa8e3614d0 : SUCCESS in 12m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1d618c5da96249e0a769b288f2b678fb : FAILURE in 8m 08s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/8486617a8830431e9deeac4f9bc6e9c3 : SUCCESS in 8m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/49ca1db42d86472dae5b860916e9a6ff : SUCCESS in 13m 09s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/b3472adedd45410ca37695c83af526c1 : SUCCESS in 14m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/95bfe46f27b549cdb02e36e5f83503d4 : FAILURE in 15m 08s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/9538260226d24d2f979a96a6c78a9257 : SUCCESS in 17m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f7646ad6578541fcaf79efdd8d1f65aa : FAILURE in 22m 58s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f4213a30c1464ab189a7d7eae093c170 : SUCCESS in 26m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/38419fdea3b144cc808c0537251ba46d : FAILURE in 10m 20s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/87c35a5fd91d4231a30fe14afc96479f : SUCCESS in 18m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3439005bdeca485aa576f5038cd02731 : FAILURE in 11m 27s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/0e9921c876c94f53bb3e03bf561fe6e6 : SUCCESS in 9m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d004cf3cd97e45469054a6d3857386e4 : SUCCESS in 17m 10s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/9db4480b67404ee599330556601a2d93 : SUCCESS in 10m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdf357118f4e41568dddce06d2e0097b : SUCCESS in 7m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e15f93d8ad0450aa391a6d28044b641 : SUCCESS in 1h 12m 44s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d60e97d90cd1494f9d010e7ab583699f : SUCCESS in 1h 28m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/38acc022a0f743a4aa97c65f5cf21db4 : SUCCESS in 51m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0b3e7e8f5c23acc293ab12071c07f7dfb48c480f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-04 08:44:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bbeab3b0d158603579b542ef8b818b0688bda6ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 10:22:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b75692848c77487ea1f2f6d063948c23 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8716aa8eae7745ffa13d7d508c9c16b2 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b175bb059263435d8ddb991b9200abaa : SUCCESS in 5m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e8cf471918a441ecbf54d1a22ee85bb3 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6521da312fa2416ca21f411b16f7620f : SUCCESS in 6m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2fbdd203bb544d07896a3d8d332b8ccf : SUCCESS in 4m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/52972148b1ad48f2801b146863cfb042 : SUCCESS in 8m 49s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/54b4ce6b9fec43368268d0addd2435d1 : SUCCESS in 9m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8e50d805071347ac820e6433b62eab1d : FAILURE in 8m 42s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b6bc6e809b004f698b7d22d6a41583be : SUCCESS in 10m 51s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/e937adf8dbdc46bead26aab2ad0d61e0 : FAILURE in 11m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7137f98c22e54c598e5ec02d36497a9b : SUCCESS in 5m 42s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b4cc9a1d652c4bbb833ca2ce876e972a : SUCCESS in 6m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8704e6fdd003478690e7ab55c34d311a : SUCCESS in 12m 31s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/0dffd2154dc241b582782a10d96e2524 : SUCCESS in 11m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a73016b536504498b61513a86192d25c : SUCCESS in 18m 53s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/7092a7c10fa54d22b6b0d19c31af0e76 : SUCCESS in 20m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/57d613e8db3249aaa394456413e738d3 : SUCCESS in 32m 48s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/366e5ec15696460ba2e484d9bb50e756 : SUCCESS in 26m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8c48e1492697408e8ea00df60fa32d53 : SUCCESS in 13m 56s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9ce8d46ae07b429fa87660f9b1f6b1ba : SUCCESS in 17m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff975710645943bf8a82f6cf176e7f61 : SUCCESS in 21m 40s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/c83412abaf0b4963b9c9a82b94642afa : SUCCESS in 6m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b056853b7363459294e24492f2485d99 : SUCCESS in 10m 06s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/65abe0b940034f1d8a7a5d867b99eefa : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4c930fb0a6342b896137429f06b2cc1 : SUCCESS in 4m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23b3e7264ee3416bb4f337960d8e89ef : SUCCESS in 1h 14m 39s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/5bbc7e5d928a4953be9398938ac2c54c : SUCCESS in 1h 32m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f335b2555df64db2b539f898c4cc62b0 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b4debdd6b287b0be5bb863aa377041b014d95b94","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-04 10:30:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed3da0bf7148341fd334d39906e5b9f01720430d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 14:05:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5502ef5dd62e433592e9cc26be5a244e : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaecb4cdb393437d92a608f380a22fcf : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b272f6e8df5d4d82b53adb1a6793a105 : SUCCESS in 5m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6c3b1cc4f9db47a388e4f4f0eea65d93 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8cbd585cf87448daabfcf275a77368a : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db20f75999b040049d1ddbd3a718d1f3 : SUCCESS in 4m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/556e21a1a328493e8ec34ecedfd9d90c : SUCCESS in 9m 35s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/46f1f7afe5a34ead8175228222b87c90 : SUCCESS in 8m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a53217d165c74ce3b169c2f688ecd456 : FAILURE in 9m 33s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5086a8783b924954aad7a2590efd37c8 : SUCCESS in 9m 51s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/bcdd25c934584bbf8edf9c4d7c46d1e3 : SUCCESS in 10m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/91b4d31e3e654a86a9508eefe7f1955f : SUCCESS in 6m 27s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2c900f350d744f88ae5217ec120bc391 : SUCCESS in 6m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d804d2b1dbbb40b8b6bd487fe0a420d2 : SUCCESS in 11m 02s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/463e5e343e2345f2a5d8f289cc68e190 : SUCCESS in 11m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1b57fa390551419697483514f97cada9 : SUCCESS in 29m 23s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/3a8d26d61fd8419ab07154813133f88e : SUCCESS in 17m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b73572a74ccb43d0be3904c933043eae : SUCCESS in 24m 19s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/60499680f4fb4ec9adeb46ea3b2f9216 : SUCCESS in 30m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/303af48af5c047a691d73f5989256904 : SUCCESS in 14m 21s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/a19e78b984004057880e494446b39ea8 : SUCCESS in 15m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/baf6771aa2eb4371a79b25d7feebbf01 : SUCCESS in 22m 08s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/b154dcb5a0d0476e849e74419f1df6ee : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8555f08e46464a8bb9f7218e8e65140c : SUCCESS in 11m 25s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/28cc827116bf4c599fb638be2b098aff : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00c2d3f6368e49e4972595623be64fe1 : SUCCESS in 4m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31281aa815bc46f79e9eb051c2718f2e : SUCCESS in 1h 08m 53s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/3c831e82dbd74bc689b4484fce5ee377 : SUCCESS in 1h 32m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a5c7f64a080844c5b6bb7d574a8dd4a1 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":1},{"id":"37c2722392c473b489956b38d52f3a11ca87e4e4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 20:46:12.000000000","message":"Patch Set 1: Code-Review-1\n\ntoo old of a change for too old of a release\n\nchange contains some potentially breaking changes to a lib that is used by projects not covered by cross gating (for not wanting to boil oceans)\n\n+--------------------------------------------------+----------------------------------------------------------+------+-------------------------------+\n| Repository                                       | Filename                                                 | Line | Text                          |\n+--------------------------------------------------+----------------------------------------------------------+------+-------------------------------+\n| airship/drydock                                  | python/requirements-lock.txt                             |   74 | tenacity\u003d\u003d5.0.2               |\n| openstack/aodh                                   | requirements.txt                                         |    5 | tenacity\u003e\u003d3.2.1  # Apache-2.0 |\n| openstack/ceilometer                             | requirements.txt                                         |   35 | tenacity\u003e\u003d4.4.0  # Apache-2.0 |\n| openstack/charm-aodh                             | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-barbican-vault                   | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-ceph-rbd-mirror                  | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-keystone-saml-mellon             | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-masakari                         | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-masakari-monitors                | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-mysql-innodb-cluster             | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-mysql-router                     | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-neutron-dynamic-routing          | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-nova-cell-controller             | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-octavia                          | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-octavia-dashboard                | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-octavia-diskimage-retrofit       | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-pacemaker-remote                 | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/charm-vault                            | test-requirements.txt                                    |   19 | tenacity         # vault      |\n| openstack/congress                               | requirements.txt                                         |   35 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/congress-tempest-plugin                | requirements.txt                                         |   12 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/designate                              | requirements.txt                                         |   33 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/heat                                   | requirements.txt                                         |   59 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/mistral                                | requirements.txt                                         |   62 | tenacity\u003e\u003d5.0.1 # Apache-2.0  |\n| openstack/murano                                 | requirements.txt                                         |   12 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/networking-generic-switch              | requirements.txt                                         |   12 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/networking-ovn                         | requirements.txt                                         |   14 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/neutron                                | requirements.txt                                         |   21 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/neutron-tempest-plugin                 | requirements.txt                                         |   16 | tenacity\u003e\u003d3.2.1 # Apache-2.0  |\n| openstack/octavia                                | requirements.txt                                         |   44 | tenacity\u003e\u003d5.0.4  # Apache-2.0 |\n| openstack/octavia-lib                            | requirements.txt                                         |    8 | tenacity\u003e\u003d5.0.2 # Apache-2.0  |\n| openstack/octavia-tempest-plugin                 | requirements.txt                                         |   19 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/oslo.messaging                         | doc/requirements.txt                                     |   13 | tenacity\u003e\u003d3.2.1 # Apache-2.0  |\n| openstack/ovn-octavia-provider                   | requirements.txt                                         |   11 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/panko                                  | requirements.txt                                         |    6 | tenacity\u003e\u003d3.1.0 # Apache-2.0  |\n| openstack/paunch                                 | requirements.txt                                         |    8 | tenacity\u003e\u003d3.2.1 # Apache-2.0  |\n| openstack/qinling                                | requirements.txt                                         |   32 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/requirements                           | global-requirements.txt                                  |  505 | tenacity  # Apache-2.0        |\n| openstack/requirements                           | openstack_requirements/tests/files/upper-constraints.txt |  402 | tenacity\u003d\u003d\u003d4.9.0              |\n| openstack/senlin                                 | requirements.txt                                         |   38 | tenacity\u003e\u003d4.9.0 # Apache-2.0  |\n| openstack/senlin-tempest-plugin                  | requirements.txt                                         |   13 | tenacity\u003e\u003d4.9.0 # Apache-2.0  |\n| openstack/taskflow                               | requirements.txt                                         |   34 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/tooz                                   | requirements.txt                                         |   10 | tenacity\u003e\u003d3.2.1  # Apache-2.0 |\n| openstack/tripleo-common                         | requirements.txt                                         |   32 | tenacity\u003e\u003d4.4.0 # Apache-2.0  |\n| openstack/trove                                  | test-requirements.txt                                    |   30 | tenacity\u003e\u003d4.9.0  # Apache-2.0 |\n| openstack/trove-tempest-plugin                   | requirements.txt                                         |   14 | tenacity\u003e\u003d5.1.1 # Apache-2.0  |\n| openstack/upstream-institute-virtual-environment | elements/upstream-training/static/tmp/requirements.txt   |  287 | tenacity\u003d\u003d5.1.1               |\n| openstack/vitrage                                | requirements.txt                                         |   56 | tenacity\u003e\u003d4.9.0               |\n| openstack/zun                                    | requirements.txt                                         |   46 | tenacity\u003e\u003d4.9.0 # Apache-2.0  |\n| x/vmware-nsx                                     | requirements.txt                                         |   10 | tenacity\u003e\u003d5.0.2 # Apache-2.0  |\n| x/vmware-nsxlib                                  | requirements.txt                                         |   10 | tenacity\u003e\u003d5.0.1 # Apache-2.0  |\n+--------------------------------------------------+----------------------------------------------------------+------+-------------------------------+","accounts_in_message":[],"_revision_number":1},{"id":"7bd58adc7d08274056a188ddad26481b0a395150","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 20:46:41.000000000","message":"Patch Set 1:\n\nhttps://github.com/jd/tenacity/compare/4.12.0...5.0.1\n\nhttps://github.com/jd/tenacity/commit/52e0181f347e1af8dbe765db654b23d48553c062","accounts_in_message":[],"_revision_number":1},{"id":"96149027c6468590bcfed83c98af7882f5577f41","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-05 21:34:37.000000000","message":"Abandoned\n\nAddressing the issue in the project in a different way.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e4dfb65c0ac23b6e36573febd87456ab67d1ab8","revisions":{"0e4dfb65c0ac23b6e36573febd87456ab67d1ab8":{"kind":"REWORK","_number":1,"created":"2020-02-27 09:04:02.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/04/710204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/710204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/710204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/710204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/710204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/710204/1"}}},"commit":{"parents":[{"commit":"77d1510feb85c38d830fbfb1ff464facb8f26333","subject":"Merge \"update constraint for keystoneauth1 to new release 3.10.1\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77d1510feb85c38d830fbfb1ff464facb8f26333"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-27 09:03:44.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-27 09:03:44.000000000","tz":120},"subject":"Bump tenacity to 5.0.1 in UC","message":"Bump tenacity to 5.0.1 in UC\n\nThis is needed to fix a bug in mistral for rocky\nsee https://review.opendev.org/#/c/710170/1\n\nChange-Id: I419edb985683c3f5c96d98833bf39dc523837e90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e4dfb65c0ac23b6e36573febd87456ab67d1ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e4dfb65c0ac23b6e36573febd87456ab67d1ab8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
