)]}'
{"id":"openstack%2Frequirements~710247","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I1b624b653253b58ee3a31d4bbe2705080729459b","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I1b624b653253b58ee3a31d4bbe2705080729459b","subject":"update constraint for ceilometer to new release 13.1.0","status":"MERGED","created":"2020-02-27 13:07:32.000000000","updated":"2020-03-05 02:28:20.000000000","submitted":"2020-03-05 02:28:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710247-1583375300785-97979107","meta_rev_id":"aa9e1ca9ac71d27ba04dc2ac270f8c6b9eee9561","_number":710247,"virtual_id_number":710247,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-05 02:28:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 22:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-04 07:01:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-04 07:01:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 22:49:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-04 07:01:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-05 02:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9ba137ae979dbc79fe3bb6d270ce7f6b8c01ca5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-27 13:07:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ef0aadb993af81080db23b270e650d6e35372f7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 22:05:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72ac9988d030e3d3a62d6d0d0c60ed9c06a5ed4f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 22:49:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30763e7e80219376992544edca841252b2757afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 22:58:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/25b403d644ed48f6bb65b10e846ab046 : FAILURE in 4m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a08f0bd8ff4e0e944cb75e771c77c2 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bcb360e0803a4cac964beffb6e924598 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b65d6c35a4d240ef9518198b2f97f4ed : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bec07448f880461eaed2ad2e0cf8f513 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc36c36903c04b46aa140af0f757071d : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15687b6af2cd467d8622ecce9f94c996 : SUCCESS in 5m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/973446f4163c4dd1b59567e8cfb2a681 : SUCCESS in 8m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3f46b31faf1a4f7782b8d5be3bfa03ca : SUCCESS in 9m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/faef9e745d7345ada3a7378f5393e339 : SUCCESS in 11m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/05d72f547e9947e09616ff7daacc76c3 : SUCCESS in 10m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3113c6d530c74685b3d1f8a4003449d2 : FAILURE in 6m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff266b86dc134601b6660494b1078a45 : SUCCESS in 6m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9e4d9a35caca4a23a8c791f60332f247 : SUCCESS in 12m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/713377e52e5b45ec88c396bcbdc3f2f5 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d6ee4f712b2410da638f7e4c0aa894f : SUCCESS in 9m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/19be3c4796ad4cdcbaba8af14e4f28ff : FAILURE in 23m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6b582becc7344d1dbd4d05c107ccfff6 : SUCCESS in 27m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/802523f30ad34b5aadb3997e796e5a11 : FAILURE in 27m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ef2e39942d1d41afb2f2d4cb31ba3670 : SUCCESS in 20m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b24be12808ac46fd889cebf41be1445e : FAILURE in 7m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/135c2e032de047098c522010f94b0af4 : SUCCESS in 15m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6b060a86b09a4a369988a671c81f7e8b : SUCCESS in 14m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a265048bb5e4efeb307ba671853b1bd : FAILURE in 7m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69a78ffda7c2488e960cb5396c729940 : SUCCESS in 7m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/11a5ce62283f4a229683033474db4235 : SUCCESS in 10m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7b4d4ef491324b14b21fea0565c719aa : SUCCESS in 11m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b6e7ba1069b14f0495d473d408e525af : SUCCESS in 1h 20m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a2c0a61524c24ec5912cd78b3152bab5 : SUCCESS in 1h 27m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e367cfb9903946d3b6a2ee56a6a380f5 : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4e85bf8df9d0797f108b0eb1b90aac029d467a52","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 23:05:52.000000000","message":"Patch Set 1:\n\nlooks consisent, but py2 is failing a few places...","accounts_in_message":[],"_revision_number":1},{"id":"d9c7e05a06707d70fa6bf69f792024782e9bd194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 17:49:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5001df73aade4585b48f278a1496e97a : FAILURE in 5m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6189e205723d410d9509d0b65d488156 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/120475ca496042ccadb23ec1d24b018e : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/445b186017d049928e7d0b7a274b9be9 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a01f040053d44e7953e1319196698cb : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3cee63a816c47eabd4ee020246d019c : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/311fb50b51ac46b5be30234915350203 : SUCCESS in 5m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/96749829bd2d40358422e412ba3604a7 : SUCCESS in 10m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a5e747900b5d45bdadf9fb7d2d0fff54 : SUCCESS in 9m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/105967e94f054d789d49c9ba86cc9b87 : SUCCESS in 12m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b57710be913a472393cac2ff1f13d0fb : SUCCESS in 11m 16s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8817d01355b942a7ba4b31999e3865d2 : FAILURE in 7m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6a1dfdd6d8c0480faf8d982cffb0d127 : SUCCESS in 9m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5855b2e163ec4c6c84420db226345ea4 : SUCCESS in 12m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3f37365c8ace4188ac11fcabb1109bf1 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff3ea55af4d64348af104a0a29a13374 : SUCCESS in 9m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0a3329680f8c4b82aaa4c2764e4f718b : FAILURE in 12m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d30bcfd132e24b3787599513a4ca85b2 : SUCCESS in 18m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2de3196e275b45e492213fa36ba2519b : FAILURE in 26m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/05deaf1b344145d9a4f9fc731eb2c671 : SUCCESS in 28m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d3e54f35e4ef422babb721330a013165 : FAILURE in 9m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f8ac2ca13f9243d58f02d227fbe994f1 : SUCCESS in 16m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/febcab94f2694cbca5e9d3c2d9a08e49 : SUCCESS in 15m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/709ffc1be4504fc9b5b019784362797c : FAILURE in 8m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69c456409edc40cf926a32954e4f82a3 : SUCCESS in 7m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a9d43bcc8d9440909c528ead6bffbb09 : SUCCESS in 12m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ee2194ba3a0435291c435478bb81990 : SUCCESS in 13m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0121727bd09347f79cf1268ae40b1690 : SUCCESS in 1h 21m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b5a3305bb75f4f8cb9f94a760f811cdd : SUCCESS in 1h 47m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/caa044912ea64605b2b737977840797c : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":1},{"id":"337626a862d26a46d71dcfcab51d5eae7671a5ef","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-02 21:13:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ec33cdaaff718a335f27668f89da1fc6b93582d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 00:45:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8475fd687a954189984cfc236f4d24ea : FAILURE in 5m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc3f3a8f0cd54e6dac59f449cd1396b8 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85b62275eca7484e9a29fc3cfe5f4209 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4426539672624e1c9cedd2c1ce99e2d2 : SUCCESS in 9m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/272e10d0fc5f47f59f48e514b282eba6 : SUCCESS in 8m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cc48cc1ab20479986bd56a526612ffd : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/856da7068a0e4e1bb0cb4b2802bce6ab : SUCCESS in 5m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5b3b4c1337ed400dbcb6cf70152602b4 : SUCCESS in 11m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fab3a25c21784417ac14777af2cc7ba3 : SUCCESS in 11m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f3163b2664814a409c093fad0e4429bc : SUCCESS in 12m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a09eb98bfbc54ef793785f52663b6cbd : SUCCESS in 12m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/54fbd6978947427596f60fd049dd1e63 : FAILURE in 8m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d65d9d726a2a4d869c903ef820ab74f6 : SUCCESS in 9m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ab6d5d564cd64d3a8920c230da6efced : SUCCESS in 13m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a1489aec42694632b4f093fde2e6626f : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09c17f93d4c84da389e4ca2e7649d5d7 : SUCCESS in 11m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1bd2fde5cf3547c58ef229a3f258bdb1 : FAILURE in 25m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7bcce1d6dc0b42aba78a61ad8454e34f : SUCCESS in 15m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/53271f6dc7f24e96ada8b02edc7ba840 : FAILURE in 27m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b90fadca6e634c7b8c603448f9f18816 : SUCCESS in 24m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b34ccc8ed12f4d74ad620d2a33a25636 : FAILURE in 9m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6aabb235e31c41ebaa58dcb1a3be7c0e : SUCCESS in 17m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f872e3ac48ce4ccf87a456ada10c6166 : SUCCESS in 16m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/938c069ec1cf47d0a93da5b0f071c6df : FAILURE in 8m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be80042767b44c479b0743a0a23e3a01 : SUCCESS in 8m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/751e57e8b9ff4aa6acafb29154eca053 : SUCCESS in 13m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3458f64d44aa49639fb8189d6b8fa66c : SUCCESS in 12m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc6ef9003ee546c385fbcf1e62b0a13b : SUCCESS in 1h 19m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b02b397613db40fc93d11df938878cca : SUCCESS in 1h 52m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9cd20206f32941d7b709b07760585522 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":1},{"id":"75782ca7170afcd61ef5a3d99b55ca7b276ce9c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 04:20:57.000000000","message":"Patch Set 1:\n\nrecheck\n\ninfra fixes?","accounts_in_message":[],"_revision_number":1},{"id":"bf8fb81c0907d43423d157137d45bb6212d2a84e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 05:49:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0c0a2a2fd0de415c9093e3fda8bc6723 : FAILURE in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cee7f670cd3d4013bce32bae092a81d9 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a8600734eef43c29664d55f3212548e : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89ef74d88ba343f084df167cdc723409 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/158aab7aca014f2da589354f93e9a3bb : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7bbe11b7f794ec6bd1e925619d2e313 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6afeb9385a945758ccf38a041d7af60 : SUCCESS in 5m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3fa1ceed2c844bd6bb2e85eff9d8a57e : SUCCESS in 7m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fdea6435547a4fad943e484d8d79b39c : SUCCESS in 7m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c86ace5c691a47b28fe1b2bef14f19ba : SUCCESS in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9058e596594b40a79a116e181d3657f5 : SUCCESS in 8m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/48637011640d490582790fb198bc89a7 : SUCCESS in 5m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/07a16ca8d17a4fa7bc7ddd50dd0145c4 : SUCCESS in 5m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/23881ac13738469f96452b8d0884cef1 : SUCCESS in 10m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/93c1fa8c37ee46aaa35ecdd086feb9d1 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7440a10cb0be4d92826334e2ee0c655b : SUCCESS in 7m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fbb7450b99fe4addba0e8cf8c7537b47 : SUCCESS in 21m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/de98ae2be620409593a3a1a75248ccd8 : SUCCESS in 12m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f8ff59e8867f4aca9e5edf5e4df8f247 : SUCCESS in 29m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ef2af91420b34f0289e9d691b853926d : SUCCESS in 28m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1af3fe28b1714814b952498d4af9f65e : SUCCESS in 13m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9147e345b32546109b95214bdd7d3ee4 : SUCCESS in 12m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2decd8688346481b9bafb2f00c3d8f8a : SUCCESS in 13m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b2b028f4c984281b56396509e43adab : SUCCESS in 17m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96dc79a25bdf4a1aa6f7698beb05fde4 : SUCCESS in 5m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/47503ec5605448e892fe69cbd28e0980 : SUCCESS in 8m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b73ba1ef28d54afb97145a3355a21f0c : SUCCESS in 10m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/460275b50f4541be942f76985d34e368 : SUCCESS in 1h 08m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/35c0b881862041b3b1c8d4c38a75f995 : SUCCESS in 1h 26m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ae9817f1a96d4d208e448b5ff56f5e51 : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":1},{"id":"119b3d0226bdde26d41ea4b26bc6671fa1295ac8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 07:01:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f3732952ab34c177af0f3a9950108805f8109f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 07:02:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb7e47e328b77e0ee71186b267343ab24fdfcaa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:18:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd5e74388b5341728308e5d36526a3b9 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f95362eb9aaf49a68b000125d2a99467 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90aa99a762394210a487a1c8121b6dd1 : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54c65d53a00a4fd3be139c65fa537f2b : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d938284b6e4a4453b3427cea5c78a6e6 : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c84ff1a1c18446a967535a8ae724215 : SUCCESS in 4m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9481c64f451a4af2a2336059732b58fe : SUCCESS in 8m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/10b494facfae478783fd9ca507942711 : SUCCESS in 8m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d90e86e1fd2947a38f209771662b522d : SUCCESS in 15m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1ad60f3e332c4d1b970dd729162c5159 : SUCCESS in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fbd5ce728f394565876d2127953b25be : SUCCESS in 6m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b66ff1ee714344fcabd3b89337ac0ed5 : RETRY_LIMIT in 3m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/48ef9d2002e74feb8c107dddcf21b7c8 : SUCCESS in 10m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/132f480405f147999b1d7339861ab559 : FAILURE in 6m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b21e6b16a6a4e9b8ec7cbfdd3cce4eb : SUCCESS in 7m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/02d1cfd85e084de281561e6cd4e0b41e : SUCCESS in 16m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/49b541b5659043df816706825e76c834 : SUCCESS in 20m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/84b76694428e4498b8a2ce0f3ea7dcb5 : SUCCESS in 26m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6952a9fc99d14182af5df4698bc4ead7 : SUCCESS in 22m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5aa65014b4d64232a256b0b555f49a5c : SUCCESS in 15m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/35a1305d89d44179a73928dd9b5041d3 : SUCCESS in 17m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f7971dd448d1433a870615fa0d1cd3a9 : SUCCESS in 17m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e8b172d7c2d42f19a3ad26985da7712 : FAILURE in 22m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1375c952ee24fb5a0976b304ae0ee56 : SUCCESS in 5m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1df575dabcec4d1f9a60b00ce187ca94 : SUCCESS in 8m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73e3c0a0c2fe496f9bd6d0a6d908a68c : SUCCESS in 11m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5213bfb1e3f44e9da168a2f556b249b9 : SUCCESS in 1h 14m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4cadb65f45494a18804488d1fb2d6cab : SUCCESS in 1h 35m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4c6fba7a2bac4dff8eb85cb9db0abb52 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b42bd69050695661a1d19d99692e16964fd0ef71","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-04 12:20:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e7ccb81e4d4fdd85f3e6cf43d5e57289a777bd6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 15:31:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/545178cd10cd49779b13c6a51331cc3c : FAILURE in 3m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9158683968254cb3b8d301307cff28c3 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63b109127e014e328812cae7ffb71cc2 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afbf69e68ecc4e2b83dcae77a0ddfe37 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15fc2c2322084306a0a880ea3e23057c : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6fc1e2499cf4edb87d639f5dad1f5bf : SUCCESS in 5m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d850b7f8759440a7879a37ab77616a72 : SUCCESS in 4m 32s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/35802723b3cd44ae9e3c2b14182275fa : SUCCESS in 8m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8630f63a356a4f96835ad4188320f279 : SUCCESS in 9m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/90b0a27f64814668b0abdd03f9e3dabc : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8edadc5970a344eabeffc6d3622d1024 : SUCCESS in 9m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8fdda7d986184ccf80f63ab4b20b7be9 : SUCCESS in 5m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9d2e5a297b134f1fb5ea106e49b67840 : SUCCESS in 6m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7055c769d57e499a8924336ae58b49ea : SUCCESS in 12m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66cfd978924a46f1bab0e3daf45282ce : FAILURE in 7m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f89e44f9cf3840d49c489d0e6d556f3a : SUCCESS in 9m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2cdc8cde127340c09b6a2f50b03c873a : SUCCESS in 23m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/996d760832274f99bf36ac2f48d60454 : SUCCESS in 23m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/aaccda95fa6b40a68024c17382751098 : SUCCESS in 27m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6752f64948e94e03bcd67d371437ab4b : SUCCESS in 27m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/61c34dbbafc5418dae2dff85dcb7579e : SUCCESS in 14m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/866daae0c0964cc58873892ad0d7c074 : SUCCESS in 16m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ab86a63f8e6e4f158e4ad6448b833ded : SUCCESS in 15m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1e73ce6af064884b8f3ab35260fdb73 : SUCCESS in 20m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b685bb62a0f74895897de5a10dcc7796 : SUCCESS in 8m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6797afa8ff0a4c76b99bd2e60525e5e4 : SUCCESS in 11m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c4df304bca8d46b7965177af47e1bdd2 : SUCCESS in 12m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a2d2cdcd8b5478c99f47a9df2235de9 : SUCCESS in 1h 05m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fcaa49bcbc1749f6a34a3479ef60ac49 : SUCCESS in 1h 45m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/784f45dd3fb74f018eb3b62ac01a3b36 : SUCCESS in 55m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0f7b19f5ddbae1463896211e714be0aae824267e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-04 15:32:33.000000000","message":"Patch Set 1:\n\nrecheck\n\nStill some infra issues that may be resolved.","accounts_in_message":[],"_revision_number":1},{"id":"91a947f53ef47bd4ce0f23cbddea8e1b8cd0a6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 20:36:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8bd49f29543845ef8dfe10fa2cba71f9 : FAILURE in 3m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2044f1a202415ea7008b6a6ebde523 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da0380d1f8a4428eb7b1c90dae16f012 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d60635088ab45549bc9715d7e23c2ae : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3641fae23a24eeb9db6bc7bc4b9a568 : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1759c1a413344869ad8264e63f52ed25 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d6699d96f73426da6d44db8c2c5b4ef : SUCCESS in 5m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9ebdd14cf9794aaabb6556cfd6f61299 : SUCCESS in 10m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7842990a7d7249a09b4c2731a088a755 : SUCCESS in 10m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b4e59aefd97d4978b1b0edfba32ccbdb : SUCCESS in 13m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5267351a433b4be78d3da57a73028fa9 : SUCCESS in 12m 44s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b76261c4707d46db95179d13715976c7 : SUCCESS in 6m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a610cf72ac264962bb7a078585b04e48 : SUCCESS in 6m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8cc66d36fb234776967f6ce906db0fd4 : SUCCESS in 11m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/87b9d5efad6141b49d3a3722a944b1b6 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53a16603b942491787807b1cfac56324 : SUCCESS in 8m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5f25b03ab5ab4617a7074143b71f4669 : SUCCESS in 20m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3da1edddd3bb4c8babf557b6c28c5371 : SUCCESS in 14m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f16d7b49349f405e864e78b3c7cd51c7 : SUCCESS in 28m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bcf75f5e509346f99669b8691f1009ef : SUCCESS in 25m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d3ffcdaaba1142edbb5ebc6dbcf2ddd1 : SUCCESS in 15m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1221766a9b6244d9a307c268c7587b94 : SUCCESS in 15m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d5df98f3a375475dbc1f199b86a05c72 : SUCCESS in 16m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2994b78e0f2445ef97378d8f4dc619b6 : SUCCESS in 27m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc24b94a62254f6da26fef836c955106 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8aedb76fcd51405fabd476a5848b2aad : SUCCESS in 9m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e2a0fa06dc484cb8926c94c01c48a630 : SUCCESS in 10m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d23f5b6c4761421f9693e222729812aa : SUCCESS in 1h 04m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/be2bd9402e134430b3d07119ba4ef408 : SUCCESS in 1h 50m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1cf8085668284df18ac686d671d57a58 : SUCCESS in 1h 06m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cc29f457dd9538ba15b4861c6b89cb457469c2cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 20:37:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2852e9aed197da62585292f4b6022dd9381c3be8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 02:28:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7824af30ae3023ca52f442c9dfae80f4a6d78c33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 02:28:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22e468278df94941bfd0841360262807 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fa6363a30eb4799b36ca73c16700f03 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c8c3d957c1645efb3b55dde3fcf22a5 : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db026d6c8df0488d8bbc333fb21477a8 : SUCCESS in 9m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/159e82b0a5394e59a9c528538ad1024f : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3eba5270dc1a4e84a99e236305c32e23 : SUCCESS in 5m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e00c538000074578a0d076a18cac606a : SUCCESS in 11m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f18a1ddf49ab40b4867e3a60d09a48a5 : SUCCESS in 10m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a88fb25c48cf4d8da5c520e55f46e4c0 : SUCCESS in 14m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e05364b9c85344e397d15d5c21c67c23 : SUCCESS in 13m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/520a1a12488b4f7fb248067ad06ba7c0 : SUCCESS in 6m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e329c78370b8446bb3fda891bc5ebaf5 : SUCCESS in 7m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f31fd9fb10ef4ddb89cf04198b6d7279 : SUCCESS in 12m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/405a2808ab8e476481acd7cc33977918 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1bca133723b4e4d80e6d9eba63c7c6b : SUCCESS in 8m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f12df7f0814e4eb9b7ad80b77dc6d2f5 : SUCCESS in 22m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/738c8f1002c24298927075604d4525f5 : SUCCESS in 25m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9ac3da218012484fb0573d3c97576301 : SUCCESS in 22m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa2c6e28ec004ac8ade2ad83993d1953 : SUCCESS in 24m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a0d6735d5a9840bc82b6f9c026ca940a : SUCCESS in 13m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8f2f775034fb40bda23f277c393ccf3a : SUCCESS in 15m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b49eacf93a2844dbbd0e91707ae69dfa : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6985fbe431fa44388bbb912a59c62310 : SUCCESS in 20m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbadade388b94719b7c70be8213a4043 : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fc63f0136bf64344ae5c98939a27bb5d : SUCCESS in 9m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/003dc2dee2484d9ca6ad0e5a6b8413e0 : SUCCESS in 10m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f8311fca89fa4576a7624e3193e4e821 : SUCCESS in 1h 11m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4dd7f69681064645a5b827b9d45f4f26 : SUCCESS in 1h 35m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fe85eff5b9f74ef3b51e2bfa3f0d1613 : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f28edb2883d9ac917536489d99b5395e792677b","revisions":{"4f28edb2883d9ac917536489d99b5395e792677b":{"kind":"REWORK","_number":1,"created":"2020-02-27 13:07:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/710247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/710247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/710247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/710247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/710247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/710247/1"}}},"commit":{"parents":[{"commit":"3fa235bb42386f704dfdd8e24e264f42fd8be386","subject":"update constraint for keystoneauth1 to new release 3.17.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fa235bb42386f704dfdd8e24e264f42fd8be386"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 13:07:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 13:07:31.000000000","tz":0},"subject":"update constraint for ceilometer to new release 13.1.0","message":"update constraint for ceilometer to new release 13.1.0\n\nChange-Id: I1b624b653253b58ee3a31d4bbe2705080729459b\nmeta:version: 13.1.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Commit: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Change-Id: I6b128ff0d22ce13e1245a4c55ce608142564207d\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f28edb2883d9ac917536489d99b5395e792677b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f28edb2883d9ac917536489d99b5395e792677b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
