)]}'
{"id":"openstack%2Frequirements~710248","triplet_id":"openstack%2Frequirements~stable%2Frocky~I3761e469036b1ed73e2d6b34286573daf105bebc","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"I3761e469036b1ed73e2d6b34286573daf105bebc","subject":"update constraint for ceilometer to new release 11.1.0","status":"MERGED","created":"2020-02-27 13:07:40.000000000","updated":"2020-03-05 02:19:27.000000000","submitted":"2020-03-05 02:19:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710248-1583374767887-c83ad78c","meta_rev_id":"a4a3920aa0ec65589ef1e2b885532ed47378f4ae","_number":710248,"virtual_id_number":710248,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-05 02:19:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 22:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-04 19:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-04 19:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 22:50:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-04 19:47:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-05 02:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b5cb9785fd8ee0646952fb5633c3f38f3465fc4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-27 13:07:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4a855e885289699ba5fa06a8d423334e1ce66a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 22:05:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d277f477599ded1b725edbdceb2081892e6e2354","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 22:50:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d9c8482046e4926b96c0819315490e4b68666161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 23:10:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/dac916ea0d56477680d9dce455ebbafa : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee40477e7904ff5ba4baa91a1718a1f : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f63c3ec64f84c3889bc70a6d16ea2c1 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/268462a342d240b6ba779094cbe0b651 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9691cbcce46a4e1e92853304432548e0 : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c7808332d1c4ac2a503b8d744c28a45 : SUCCESS in 5m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b01ee69b8e794f4c855d3f6ed0f38636 : SUCCESS in 9m 10s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/5a54a2c164434d61a8e61d28526ed9d1 : SUCCESS in 9m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70fe8ad154f5420abd63b283f05c669b : FAILURE in 8m 39s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6c478ca050ec4097bbb6f1a0216df829 : SUCCESS in 10m 26s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/dc33672fb5ad41c5957639336c6f8d2c : SUCCESS in 12m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b39bc28cf52d4f808d8dc54e5d5ef7a2 : FAILURE in 10m 28s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2ba7c37dfc0545059ae532f3406cd6b2 : SUCCESS in 7m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5d89a922c822471581bcd6a03c0b3ea8 : SUCCESS in 13m 45s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/5c2beb26561c4f0abbebfbb2a5b93a67 : SUCCESS in 12m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9ce97f9487b347f78764741c165ad9a1 : FAILURE in 19m 43s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/1741d0b862f4439188d0ced40a600e56 : SUCCESS in 21m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a49740e058244029a58291f71045335f : FAILURE in 29m 43s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/159d3971d71d4dedbc5f8f54c099e6f1 : SUCCESS in 30m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f29b8c5a17e0424eb30f5a5155bc673b : FAILURE in 7m 34s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/39e12b4801c348d9b14fdd8e9d0a58b9 : SUCCESS in 14m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3bb73416d20a4030bfe82d7244457e90 : FAILURE in 6m 49s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/2efe41498e5c45caa6be0f6b9ac56ddc : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e116d4581c73452aa6fa71ab7b10337f : SUCCESS in 15m 37s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e20cc971584c47c0a07420d4260fdeea : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a5d1e0083964842a94630fc24b6bcff : SUCCESS in 8m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/43a95d8663af4767a5df97324dbd5d32 : SUCCESS in 1h 05m 24s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/03f0327a8fde4d988b77df856c6b1d73 : SUCCESS in 1h 29m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/337eb3285c434957831b1822b1700152 : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4b19a6436d0d749d46b04d1f8d8110b56690d956","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-02 21:13:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8118b4b085cd65d472ca728c44517d1307d542e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 00:32:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/31cca3053fc04af1be6e9ce5404e5de8 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d918e6e1848f4213ad2a14d2c7c47200 : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ebe81dbbb671485b998383ffa0ca3af1 : SUCCESS in 8m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db7869bcef6b4509ac17a50bd95c644c : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb855f961cf64cd693f72cb3df673887 : SUCCESS in 6m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22ffd78283084c0e9a9ef3bdfc2a40ac : SUCCESS in 9m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b8fd89b862fa452794192c6759caeb5e : SUCCESS in 16m 35s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/a3a14f61c4094432ac2135bade611e81 : SUCCESS in 15m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b61d13464b344370bc1c251940bd492b : FAILURE in 10m 57s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/977cbbbdcd4c48a3ad87da96ca1ee13c : SUCCESS in 11m 42s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/807021b6cdfe4f6e8f6c75208603396f : SUCCESS in 12m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6366c476e2de475ab86663f1f5a0eefe : FAILURE in 8m 30s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/f69c8339e45e4f60ad4965d4edc5ce4d : SUCCESS in 8m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3f9cb7770368406eb6dde11f0d66b061 : SUCCESS in 12m 48s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/9311087531d34a36b4e0374a308dbd66 : SUCCESS in 12m 14s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2a9e4990e53249b6b7c83dc377abd14a : FAILURE in 13m 48s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/1d5aff78a31c4b0fa088d1c4fb4ba659 : SUCCESS in 15m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/379c4fdaa58043738b2c46549ed5af45 : FAILURE in 29m 17s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ec67245bbf244058b375f2e06c64c32b : SUCCESS in 27m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0545585320844e1bb41dbdcd0dc26315 : FAILURE in 9m 11s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/634e20c24b5f46f5b6486b12d773e5ca : SUCCESS in 17m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a8fc6b0c299458b80341e65cd7998d3 : FAILURE in 7m 25s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/10e1ec7e1b4941748ed9c1fc34732d66 : SUCCESS in 7m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0f6aadd4862c45e6818b997411773deb : SUCCESS in 12m 26s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/86e0a042fa90411694f8bf5f842e9039 : SUCCESS in 8m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/330f62478db94fe59a0e2cd50e2b9f1d : SUCCESS in 6m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0a1f21b07324f54892255ef9b69b87d : SUCCESS in 1h 09m 37s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/2152b745952e402caf180f474bf63a12 : SUCCESS in 1h 36m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f9897781a5b34dd1b48f3b1c2204a6fb : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":1},{"id":"884db3a1085816cac687d4c5cce0e6ef163f8159","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 07:02:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76247158b54cf66d19478af269170383be9c6378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 08:43:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/15dd098304cb4ea690999444bb225a91 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c319fe4cf9434d559cffe96a350cc169 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/441ba79d2c584c43beff76b9dcbe0fd9 : SUCCESS in 4m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d1f99efd5e7249638417978b881b0dcb : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c35aaede7ec244eaa1300c4e554d288d : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b12f7d6335b2403ab0b417f477d4e0c8 : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b771dc4ca3e447bda085807c6c1d0388 : SUCCESS in 7m 49s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/c913ad7af86e47c385cac560142730d5 : SUCCESS in 9m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0f522829d57042b48d5a50c91634d449 : FAILURE in 11m 59s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a5ffcf3ae39c4f49bc5cde876bbf3dae : SUCCESS in 9m 18s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/f1766014b7c14726b98e7b48ccca2927 : SUCCESS in 10m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cf72cb9ce19348a1aebec5150da56e5b : SUCCESS in 6m 14s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/8fa42e136a6e4b51a50ec50b966fb47f : SUCCESS in 6m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4ecb3eb62e0b40af8193647b1efac13e : SUCCESS in 12m 03s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/7604547caa3b4ddd904aea03c4dd520f : SUCCESS in 10m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/affb999dedc5424dbb85b01d9bb1526f : SUCCESS in 17m 33s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/8e901135dc3f434aa492b450c63c9912 : SUCCESS in 21m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1b2d9b1e6c574166bbf6b4bd5f5a516f : SUCCESS in 25m 18s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/9ac7e986b31a456b9ded1786f8f65a91 : SUCCESS in 26m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/edf830b31c684f54a209d29ab9eabc35 : SUCCESS in 13m 17s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9ebb6d7e294e479681f8393999b45b66 : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b02641a541b146ff86ea2894bdde0184 : FAILURE in 21m 16s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/3efa44d9817f4ac295457e692b17000c : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0354e24253644698a34c83f11b2dc3c9 : SUCCESS in 11m 24s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a78082864e1846ecafccc3a103205631 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc5c7f1440214dd992f1de2f09b6db2a : SUCCESS in 4m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0771b13cbab49c794a9ebd449faa028 : SUCCESS in 1h 09m 48s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/763004741b744772801424f7ccf74fa2 : SUCCESS in 1h 35m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c0ee047fb5534831af7422638d86640a : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":1},{"id":"51ac16200506678cd6f01f736bd493e9d0432c03","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 14:38:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc264cca05c62003ac27745de37df93bc10566f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 18:36:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/10857555a9914399b4cfba09ccbd77b2 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b50517a25354cc09ebc1088680026f1 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2daf62d2189348238f1880965ae8a102 : SUCCESS in 8m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6063dcb367504b57a8ea769dae11c4b4 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f3e3332721041188e439cc416521268 : SUCCESS in 6m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac27b292185443018134c303fa14a420 : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e92ae682812647d9b67865c8a166fcc9 : SUCCESS in 10m 21s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/a7d2c42d98b841fba4c26c479c7abf1b : SUCCESS in 9m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a467fbb920445c1b5c94cfbc482383d : FAILURE in 9m 35s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/44c18bb46a3e4b749c0072351b5e8451 : SUCCESS in 11m 49s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b79b6ed7668b4bc29b80f23efacd3624 : SUCCESS in 12m 45s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/24efb627277b45229170988f3fa53f0b : SUCCESS in 6m 03s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b5aa635247244d3e9387cc7e0e77367a : SUCCESS in 6m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eaf4e102ecb643cd8fca5f87b373fed7 : SUCCESS in 11m 19s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/70670c6939fd44fca293cf429fa567be : SUCCESS in 10m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6e0b1cecc9e147c2b73db3e2b2cad49f : SUCCESS in 21m 25s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/14bd3e7fa6ea4b4abf2cc4087743f58e : SUCCESS in 21m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c1c4fe4a4a5d4e599bd8910f84becf2d : SUCCESS in 30m 14s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e4c20aec39ec4adc9cd16af6ce440d8d : SUCCESS in 30m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/646fdedaa58541398887508f22e0bb2e : SUCCESS in 17m 45s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f1e8e983d04e4576835b25fdba7b7cc7 : SUCCESS in 15m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/153ed1e5d5b24b358faf70154db9a82d : SUCCESS in 19m 59s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/1fbed91db6614b3f8ff0f593fe861b38 : SUCCESS in 7m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d2fb7a66d7294fd2b78ec30b693e6fe4 : SUCCESS in 12m 04s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/1eb59591953c49ff813dadd57ece3944 : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b35afce78ba24f769e70898b8ccc05a5 : SUCCESS in 6m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5577c137aed4a94b79dc6dd93506704 : SUCCESS in 1h 17m 59s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/bd9140b1d809494eaff0eb1ec6977544 : SUCCESS in 1h 42m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2c97751f096146549875f305d58b40f9 : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"46b53a1a6ec1e7356350abc3d99c37b8f102fbe2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 19:47:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce0c6ca03a66d425ed360381a8a52b8ad81d775d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 19:48:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6814d79826641ce8a4d582e481d31360641b5c74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 02:19:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c6ee11c2950c9d99756498ea78b4bc6ebb4b7316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 02:19:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88079eacb8eb46b7a4a8a1eab0114728 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/768f837000564945bc2cc30111d08f85 : SUCCESS in 7m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/650d99caefa246d0a66089c9cdf54cef : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/619820b1e45a4a93b6b734e9d693823e : SUCCESS in 6m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4ffd134a1904657bb767d92af59b060 : SUCCESS in 6m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7cbd5b544f5a421a9d3e29f38d5ef58a : SUCCESS in 10m 03s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/ac479c528309401d9f29f985709b7a40 : SUCCESS in 8m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5cd319389c814194a3d721ee63e106bb : SUCCESS in 9m 35s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b4839206e2474d449335f4e5e50c019c : SUCCESS in 10m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6542561078524fd4a03e6d72e9245f4d : SUCCESS in 9m 04s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/7dca7951cb9240e184cdf904b4b96994 : SUCCESS in 6m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0ca58571ad494ee4b404802585efe5d5 : SUCCESS in 10m 27s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a4dec44901e8421dbfda5f7a2534a9a9 : SUCCESS in 11m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/93671156d67e45b59867fab6803f50bc : SUCCESS in 18m 40s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/9688e4ea084946838dd5a233218f2143 : SUCCESS in 30m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0bb73c22831b49e79c344eb3432368bb : SUCCESS in 22m 50s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/05125d2883d44ca9b6d9f0f884f61053 : SUCCESS in 33m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1d3a7ab9f7af4ffa90741ba12f0bf046 : SUCCESS in 15m 04s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/dd98f0cb7d7243489cea96228212689e : SUCCESS in 18m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e9086eeea844db5a8ac0c47c399f5ff : SUCCESS in 19m 33s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/c279ff5be650421b9fce72cb7c113b7c : SUCCESS in 6m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0692643b678340829044d2c7ee7fe9fa : SUCCESS in 9m 16s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/2f6b3749e3104ecb932c968070e22172 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04ec83a90dba45cd882355f283dfcf79 : SUCCESS in 4m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/432f8aa9e0654d0fa1e548002ac485d9 : SUCCESS in 1h 10m 04s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e9ac876b394c4e9db90ed6998554996a : SUCCESS in 1h 26m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/882e1219ce9a48838af2c69e686277a0 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d85deef8b5983a70e0749877ac4f36eb3541308e","revisions":{"d85deef8b5983a70e0749877ac4f36eb3541308e":{"kind":"REWORK","_number":1,"created":"2020-02-27 13:07:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/710248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/710248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/710248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/710248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/710248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/710248/1"}}},"commit":{"parents":[{"commit":"568bc433d3bee63ea052b3ca352b4cf33164f021","subject":"update constraint for python-watcherclient to new release 2.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/568bc433d3bee63ea052b3ca352b4cf33164f021"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 13:07:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 13:07:39.000000000","tz":0},"subject":"update constraint for ceilometer to new release 11.1.0","message":"update constraint for ceilometer to new release 11.1.0\n\nChange-Id: I3761e469036b1ed73e2d6b34286573daf105bebc\nmeta:version: 11.1.0\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Commit: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Change-Id: I6b128ff0d22ce13e1245a4c55ce608142564207d\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d85deef8b5983a70e0749877ac4f36eb3541308e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d85deef8b5983a70e0749877ac4f36eb3541308e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
