)]}'
{"id":"openstack%2Frequirements~710254","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ieb363dca9cae2b40e1c8aa572ed778a2b6991997","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ieb363dca9cae2b40e1c8aa572ed778a2b6991997","subject":"update constraint for ceilometer to new release 12.1.0","status":"MERGED","created":"2020-02-27 13:35:20.000000000","updated":"2020-03-04 21:46:00.000000000","submitted":"2020-03-04 21:46:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710254-1583358360208-dc82b96e","meta_rev_id":"9351d8e0402916d6ef7c3157a59fb7f25586656e","_number":710254,"virtual_id_number":710254,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 21:46:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 22:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-04 14:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-04 14:37:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 22:50:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-04 14:37:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-04 21:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75e065450d8311e3164b508febbbe6ba785febc3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-27 13:35:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34cbb992c0ec167a6952ae284d84076b97475562","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 22:05:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6aaf3b65143d16f3d11147153a2de2367f97b1ff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 22:50:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"abbce51e34e71b344bf68096f2bdf098718e15b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 23:45:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/dfc76dddfd584568bbf10025ccb23773 : FAILURE in 5m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54799edf038f4fd98545d72e20e52afc : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e4c9223c7854ea9b1e633c09adbdf5a : SUCCESS in 5m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/61f15b9a2cd6432aadeec746f3efc51e : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6838d4a8949436084ce6af0d746a0df : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ca909c3fc774710bd28107de7ad793b : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/646e7a28f7c64b98b75af7ca6e22e665 : SUCCESS in 6m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d2f51886e1854e06bc3d17fc1952ae44 : SUCCESS in 11m 10s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/25cea495d6b240ffa09cb73795eaf6a1 : SUCCESS in 9m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5cf8d6d8473547afa5bd3f0818d7a179 : SUCCESS in 11m 10s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e722b1a105ab48b6ab565036cd82eee0 : SUCCESS in 10m 12s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b1bedc4d1cda46909ad7c48ef7066200 : SUCCESS in 12m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/23183c53684f4c4d836e999b0075725b : FAILURE in 5m 55s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/bf67efcb163f4e2cbefebf4158c500b7 : SUCCESS in 5m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5bc3a905a53947adbf275f5c7939682e : SUCCESS in 11m 08s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/33a5bc785b8d48eea5a2630e8c1d6bfc : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/877db722117e465395bc858422ced89e : SUCCESS in 8m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bbbb463c753e4c6d831f892bee08ed76 : FAILURE in 10m 02s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5906e6a4b0ce4c538840be950a82dcf2 : SUCCESS in 27m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d74787cea4ce47e281bdae596e2c4e90 : FAILURE in 31m 20s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f05addbd3d184d1aa22d09df989eab32 : SUCCESS in 24m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/08ef6df3618b42e6872fa8d7976b45ae : FAILURE in 7m 48s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/458b6adf22cc464fb653674db9555761 : SUCCESS in 15m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8aea9ae092f74ce18c23565e6bdcf2cc : FAILURE in 8m 25s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/d17d766c01bf4eb880e8b50e829dc44e : SUCCESS in 8m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f91212a9399248ef9eb1eede0f93f406 : SUCCESS in 10m 29s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8992bb73c4dd4d80b6a6d1b576a753b4 : SUCCESS in 9m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4fda952fca904e5fb2f6e2ef0ace4e17 : SUCCESS in 1h 12m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/92d7dec4031f49bebd2ae6cce3a94598 : SUCCESS in 1h 37m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/649d7e3be55f4fc184e9c04e931d8ec1 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":1},{"id":"92574dd710607f696e93be2700e310b25cfc789c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-02 21:13:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fdfa9e88c275e73cafeaf77f6cfbed028e045399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 00:31:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/22321ff16a074701a420614c1fb35da9 : FAILURE in 6m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1af6647b05c646a0a46a86114d4da459 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99de69b980fd4a7d92490128d86442f1 : SUCCESS in 8m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/00cc96d1d2474ac88225c1942dff705d : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bf0014f08bd4572b18f8feb50378642 : SUCCESS in 8m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7595d5d594a44dd91bc2c8441ebac33 : SUCCESS in 7m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9af6714e0fa414da1770d9c27abd8e7 : SUCCESS in 7m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7703ab7b9be748af84a7f55d35ea5015 : SUCCESS in 11m 53s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/a80e57e06db24637a63797dda99668a8 : SUCCESS in 22m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/115cf4562a4e4864a741c45382ef9575 : SUCCESS in 13m 04s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c3d18a8816714acf89c9def15d619225 : SUCCESS in 12m 39s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/299462d1fae24561ba83067beed4ef29 : SUCCESS in 12m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3121586c1f6544d59851891e804f4342 : FAILURE in 7m 27s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/deedca8002364194a9e010c9b977ea7c : SUCCESS in 7m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5b5de77a05724bb2be72291e81db8ad2 : SUCCESS in 15m 01s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8b4c31e8b93345028520cb6182a8b9b8 : SUCCESS in 13m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e85d6e5e39ff484ea0e7d3debfa41dfd : SUCCESS in 10m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8db4620f22204043a8e8a1b53ad84464 : FAILURE in 11m 55s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f58b3335ac404514af457df90dc52225 : SUCCESS in 33m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7225124ea44a411f8e9a2bf482f68113 : FAILURE in 28m 46s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5e50b8da92b04ad2a817d7514d11f5ce : SUCCESS in 28m 57s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/63a953b0ce9c4a51aa7614dc69f79ef6 : FAILURE in 8m 13s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1ac2bc7390e44de19fee3daf67ac6548 : SUCCESS in 16m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db14f3c865ca4f92b7f4ce7874df11fe : FAILURE in 8m 10s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/bcd766838361449ea517b061f3ea6fd3 : SUCCESS in 8m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/82edeb8902594e80ad9253c60a74428a : SUCCESS in 11m 40s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/6a4edb1808a2494891fac625eabb7746 : FAILURE in 13m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51a0286afb124896ad1938ab5997aab2 : SUCCESS in 1h 26m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8e3d1e5905ce43bd856cfcaa9d259015 : SUCCESS in 1h 34m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0fc438c3157341c4942626baab4e0d44 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e3630c737216b10c7f633dd99f1e0dbcf9178fd5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 07:02:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7fb3ea15437f958d84cee839ca624842e8494fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 08:28:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7c493d95d5b84625ba848ecc275f6be7 : FAILURE in 4m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e54822e1b24f8d952a5357609629f5 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0bcbd031ca244425884518378c81f0bb : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/090cdd7829a84c0e880e3a020b5b0d27 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39588ea9c0244df3b5357f56f3924155 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ce29f7a9e7647fda32599e7db0f87d2 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/804136cf2a554a6388f6dca615238c59 : SUCCESS in 4m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bbf8b770500345968c30d781609d43b7 : SUCCESS in 7m 17s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/206ad65bb6e64438b43cfc704627a72a : SUCCESS in 8m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ec88fedf3b354267b8da8536ab47f513 : SUCCESS in 7m 58s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/47394dd1691a40279de197bccc5176ca : SUCCESS in 11m 04s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7ff7a77bf16f4016897eeb71c3c46686 : SUCCESS in 11m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/adbd899361de41f7b5f8163abd181a76 : SUCCESS in 5m 15s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d3e2d28f00d547b9a7d83a08c3b38d4d : SUCCESS in 6m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1e96151604054af8938d85de7985726f : SUCCESS in 12m 26s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e302b94b4dbe41bab718bfb645a39c61 : SUCCESS in 14m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cbf8978ae8fa4b13adfb3b5c5685f5f4 : SUCCESS in 7m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/549cabc46e7f44db986c13d1fac5b013 : SUCCESS in 19m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e544bfdd59fd415286b79d081a97b1c3 : SUCCESS in 28m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/922a36c9f78d49ed905ebad06b9a3dba : SUCCESS in 22m 36s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e859235543f7451ca9366f1151c6f926 : SUCCESS in 26m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/02688d15d0ba4cf8a7079c9494a87179 : SUCCESS in 14m 09s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/b19e893e516945c58c646af0c9ac7014 : SUCCESS in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9bc661a9a63a49e0b432b494c776b245 : SUCCESS in 18m 40s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e38e1ba7d02b44e480aab6c7ed072b4c : SUCCESS in 8m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/03edef9182734d37b60b7d4801f80178 : SUCCESS in 11m 19s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a844a100d2334920a16589868d0c70d9 : SUCCESS in 10m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c83567801a748ceb2613464b93ae770 : SUCCESS in 1h 11m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6c0526d1bc284994bc95fbe19d8a5588 : SUCCESS in 1h 24m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c7894bd1c9a048c6a54c57aa9c02338c : SUCCESS in 1h 03m 53s","accounts_in_message":[],"_revision_number":1},{"id":"db2bc62a4d6d51a772c7c1eef3a2ed66015cb229","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 14:37:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a5005e1d69f0e5a21d6ae00d3da2d0464a60da48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 14:38:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ea1f3bd62eee422270a2e9cf0ff5907292fe33e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 21:46:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c00f08eaf767e519de6eeaa030336b619497fbdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 21:46:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03db1bac4c2e4b26b9ca4fd75f6e4815 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ce64f8301fa48358c42e13230707408 : SUCCESS in 6m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ff1e44275d864d35981495ef6278c5e7 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e021e7812c86451d9361fdf5e92800cd : SUCCESS in 8m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b588deff9c941eaab02bc9ba83427de : SUCCESS in 6m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7414dc13fd8b4aabb2c0b042f650c372 : SUCCESS in 8m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a95c47709ac2427fbb0b97ed46feac55 : SUCCESS in 14m 18s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/2df5443c1799439d9c62b95cf54340fd : SUCCESS in 13m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/89a8f4f6eb014790ac51ad4910aca116 : SUCCESS in 11m 26s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/4e2ec2b0883248f68b7338f2be95ee52 : SUCCESS in 10m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a2e1cf4aba8746e29fceee7d4bbf6c1d : SUCCESS in 6m 17s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2e46600158964bbc9cf0890510a50ee9 : SUCCESS in 5m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fe1421ae0f2d4d4288ce8c7ded3552f2 : SUCCESS in 13m 22s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/5d585e5709a94056823c577044acaf0e : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb3c909182294faa9792c92719f10a23 : SUCCESS in 7m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/78ddce0f74b644e58105364161047ef0 : SUCCESS in 32m 43s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a8e6c6651095496d80df02af218fc5b4 : SUCCESS in 27m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2061ea67e0174017b3684aabc75e4b29 : SUCCESS in 35m 35s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/851c29f93c5247cea587144bcb1c9cb5 : SUCCESS in 36m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1003a6919eb4416189dd755ecf993f11 : SUCCESS in 20m 43s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/eea1236e591e452a90a5877202b70dd6 : SUCCESS in 13m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/11678ace6b1c473fbaa4bc028d75e21a : SUCCESS in 29m 30s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/2853f3446f794171bed710be5c8418ef : SUCCESS in 9m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/33c305580c4844088ce1dbace84992b9 : SUCCESS in 8m 44s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/05666ca5a1654bdbad4f32acb87d4737 : SUCCESS in 8m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb68493bd72e4c8ca5cef60782b465ca : SUCCESS in 1h 28m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/36bdcc386385479b919d55641e0809a0 : SUCCESS in 1h 28m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5fa20e5037e240ed93ea4a6defdca321 : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58120c0b6b2ad1a07750b7eebdc5da8de6d527e9","revisions":{"58120c0b6b2ad1a07750b7eebdc5da8de6d527e9":{"kind":"REWORK","_number":1,"created":"2020-02-27 13:35:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/710254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/710254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/710254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/710254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/710254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/710254/1"}}},"commit":{"parents":[{"commit":"e796f0333c6ed32dc6184e3d7413ffd1ac66f91e","subject":"update constraint for neutron to new release 14.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e796f0333c6ed32dc6184e3d7413ffd1ac66f91e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 13:35:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 13:35:18.000000000","tz":0},"subject":"update constraint for ceilometer to new release 12.1.0","message":"update constraint for ceilometer to new release 12.1.0\n\nChange-Id: Ieb363dca9cae2b40e1c8aa572ed778a2b6991997\nmeta:version: 12.1.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Commit: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Change-Id: I6b128ff0d22ce13e1245a4c55ce608142564207d\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58120c0b6b2ad1a07750b7eebdc5da8de6d527e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58120c0b6b2ad1a07750b7eebdc5da8de6d527e9"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
