)]}'
{"id":"openstack%2Frequirements~710273","triplet_id":"openstack%2Frequirements~master~If7a28baaaa7965c784c751e0e34889c58403fd9d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If7a28baaaa7965c784c751e0e34889c58403fd9d","subject":"update constraint for os-win to new release 5.0.1","status":"MERGED","created":"2020-02-27 14:14:37.000000000","updated":"2020-02-28 05:05:06.000000000","submitted":"2020-02-28 05:02:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710273-1582866165480-d82724f5","meta_rev_id":"9f8f911de2519fbbc0551da0f0877a4289389b43","_number":710273,"virtual_id_number":710273,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 05:02:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 22:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-28 03:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-28 03:19:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 22:49:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-28 03:19:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-28 05:02:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd8381c463c6d8042871d6352f433b3ec9609d85","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-27 14:14:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2442cebafe7b5b688c4f943e1f1808ad37b5a28","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 22:05:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0373a7f40e6fc8f2710e453a71594530acdb7244","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 22:49:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f6454d7ebf9327e87e8a2280ccbe41d98b57e70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 23:36:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129a01554c1444dbbc2795e508236694 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a160bd9375e74ee5a163c08453a28654 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f01598a562d041e9bc0a88f0a171e9a2 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85e99974857e428fac6052bbaf222d2e : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8da0578315b447b1a6e64ec1fa0c2fa5 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0b69f40e9d549898d8cd53c1596e481 : SUCCESS in 5m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17ad7ccc3fed44009f1a5e6d27da1490 : SUCCESS in 6m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc0556d59fe94749b20f6dce73da4769 : SUCCESS in 9m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/57b271c281204396b8f64f7306510535 : SUCCESS in 9m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e89cc339cc1446b5a1ec4219542bc226 : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f2e2d2979cac49438adac7363c57c9ab : SUCCESS in 14m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a7b569829324d10a5fdf6066366a02e : SUCCESS in 10m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/21891671b34c4ed0baac2720602e1b76 : SUCCESS in 24m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/08fc8958309344cf9f403a7569f1ecc5 : SUCCESS in 36m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a29dc3b735e04e74b0a96621a0d79547 : SUCCESS in 22m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a08ed21bca041d382a57a71faa444d7 : SUCCESS in 23m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ca1f13938ea4dcaac47b7896388d6a8 : FAILURE in 24m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6a5a81c0ac354499951eccdff6300c40 : SUCCESS in 7m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/33be2ca39fe540f8a98acf94d1e49540 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8fd6041f0c846d1964c3a784c5f69f8 : SUCCESS in 8m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c83ac564462f44328e261ffd2e8fdcf3 : SUCCESS in 14m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e9c1fb58795c4ba695efd28d60bfd694 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7705af90141b41669186c89ee1de3bb2 : SUCCESS in 58m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf24eafb43704408aebfc29935a8401c : SUCCESS in 1h 08m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/242f1d92f6004f0186d3acded385b440 : SUCCESS in 1h 14m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b4eeac43bc7c4cbf97daebe67298546b : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9b40e0797f1471e8ce370f287cf899b5b50a9f7e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 23:38:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70780657b4f9de7c0c9c24d741f89469bd691005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 03:17:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa2ae1b183a4a969cbda5f65d8d3eb9 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fcf2a79366143ac9a921742a4ca213d : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be773af0d0c141859852d2f9df8f980b : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/576c6c9cec0e4280b18c516bc0454a2a : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f8d1039c82f49ce8b1a03aea8095277 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b39b906b596b466a901cc38bbcbb663f : SUCCESS in 6m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cebc3af3b8f04131bcad2289e28c9787 : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/272aa2a102554b73802c7a16d48df06b : SUCCESS in 7m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cac63dcedc28462ab311824fdc781466 : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/683595f0e3804c7b9097df98e8576aaf : SUCCESS in 7m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e65361b954a64cdea601f68bfa1dde6a : SUCCESS in 13m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e718377372c43c1950a2bc8f4e7829c : SUCCESS in 14m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/11190c4fbea44cecbfce77939773df4b : SUCCESS in 21m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5ba24c4348ba49e096f70f40b3acce37 : SUCCESS in 22m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0063f35817574006a92f83b13d1631ff : SUCCESS in 15m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6982047c5bcb4a7996669d18244b40b2 : SUCCESS in 15m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96dc539a1cc94eb893188e89452fc113 : SUCCESS in 19m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/62570fff3e984d55a5d0e91be7ed8211 : SUCCESS in 5m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/60a61ebc3e3f4b4abf25a68ea860196a : SUCCESS in 9m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7383b020f34e45ab960fc63a4521cbd6 : SUCCESS in 8m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9f97cc6e141d467ab9154349a31b92fa : SUCCESS in 11m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c9ce57beab2b4571901e474e8931db07 : SUCCESS in 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e10454faea204705a2df86cb9563afdd : SUCCESS in 1h 08m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19799895b30748c2b45fb8971b21f872 : SUCCESS in 1h 21m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1aca24f789f047b99971d9c8afce4458 : SUCCESS in 1h 29m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8344fd35275347ac980a5250ae7155b3 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e2215ff5aefa37c4d4d99c51be4e4769e1ae7e3a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-28 03:19:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2a91ef98cc671e063707382ff0c08798b50a02a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 03:19:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8fef2224abd1c7fdb6c5813cc7ee92eea81b41d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 05:02:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe267746cb7417a96ed2584a59bdec3 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3de72a3e0fb45ad9086aa2ae30b0f33 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a13f6a464d8342db904b258a0a757beb : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c33d0b0be2241efb59064ba5a76c147 : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1ad82685a6d4139bf4f59a3878965a5 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1cce2d321a3b4a1b9475607b0877b574 : SUCCESS in 5m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f75cc8d7e2d5426b975e5072785e629e : SUCCESS in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6876e7e907e94397889d1763d187f49c : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e341aa074e3b4664b341393ee19571fe : SUCCESS in 6m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1d8af5bee63f45c59073206cdc763785 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d3ba9bb405f4f4c9700f6ca7a383e12 : SUCCESS in 9m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fbb424c9f84f4f8c884a5444c9387e87 : SUCCESS in 14m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/085ca45731fe4d80953f7c498b4f0ac8 : SUCCESS in 24m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c88a2b6a9fd24b08a111730773f891d5 : SUCCESS in 14m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/01bcc8e794eb4c3a9559b76ea90d8f73 : SUCCESS in 14m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9860837f11641a5b53fbb5e7e3b1f6d : SUCCESS in 18m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9290f59208b14483a196d5f179ac5419 : SUCCESS in 6m 10s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1212c4f7599e42a394e065a4f04ee848 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca246eec72104d0d96207f065cdb28b4 : SUCCESS in 8m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/86c15f0922fe46cd930b5989c820da8c : SUCCESS in 10m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/86a3c926ec9f4c27b92ea4a5cabd093f : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/022e72b4304b4850947cbd9742727dfd : SUCCESS in 59m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4d7975fbfbd4ce89459c36617d33128 : SUCCESS in 1h 10m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/544493aa8dec4645a2753aa02920e265 : SUCCESS in 1h 34m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/db8193f7c8a4432091172b4648691445 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"889d2a3e712e041a716dce6b92c573059a4e77ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 05:02:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ef520bcacf4041f55b48c9c94294ce53e810b7e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 05:05:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85cf8c9043fa42cd9c54c09c7eb7d1e8 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2658f04ed1ce7b6d1cce6aecb9e892012d4a98eb","revisions":{"2658f04ed1ce7b6d1cce6aecb9e892012d4a98eb":{"kind":"REWORK","_number":1,"created":"2020-02-27 14:14:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/710273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/710273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/710273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/710273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/710273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/710273/1"}}},"commit":{"parents":[{"commit":"860ebba65255df05cd55501da7d443c45193a3f5","subject":"Merge \"update constraint for oslo.rootwrap to new release 6.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/860ebba65255df05cd55501da7d443c45193a3f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 14:14:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 14:14:36.000000000","tz":0},"subject":"update constraint for os-win to new release 5.0.1","message":"update constraint for os-win to new release 5.0.1\n\nChange-Id: If7a28baaaa7965c784c751e0e34889c58403fd9d\nmeta:version: 5.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta:release:Commit: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta:release:Change-Id: I1abb5893a60c33a34cc2e365b1102d7556c293db\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2658f04ed1ce7b6d1cce6aecb9e892012d4a98eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2658f04ed1ce7b6d1cce6aecb9e892012d4a98eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
