)]}'
{"id":"openstack%2Frequirements~710309","triplet_id":"openstack%2Frequirements~master~I26dd1ca1b06b487248276b11a68f9c6c1067070e","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I26dd1ca1b06b487248276b11a68f9c6c1067070e","subject":"Manual run of generate-constraints","status":"MERGED","created":"2020-02-27 16:19:18.000000000","updated":"2020-02-28 04:44:59.000000000","submitted":"2020-02-28 04:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710309-1582864976956-577aa814","meta_rev_id":"9f1be8adf5cb3d77066e856a766ebe8d5226887d","_number":710309,"virtual_id_number":710309,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-27 22:05:11.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 04:42:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 17:10:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-27 17:10:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 17:10:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-28 04:42:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89e80b33b57555dae930c4fb02238cc756dc1854","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 16:19:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23e6226e87557271a47e19bef2f43ed71b8d81c4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 17:10:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbb1cedf930230e75524aa6431f03c7f608b3884","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 22:05:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"14c0b7b52c2205b247579ed8e6b9ac4c55e9d459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 01:27:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b30525f3a8674d20a5f66efbc1b5d13f : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db519b513e154837953e874e014b556b : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8425217ca9b420882623adab4a797df : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a30d2d2b67bd4a789f84106f2b4cedd8 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7c5415db6904be385e86202cc357c1f : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d2a210152f141df9ef0687916a517fa : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0205620acf8a41dca8b52328bb98eeb6 : SUCCESS in 6m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/607a5bdc893c426389c37e63d490fea3 : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b8a64082fb1463f89cee520f5e7edbd : SUCCESS in 9m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/103bac4a20ef4397976e3de474d53b50 : SUCCESS in 6m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3b2216f32a7041fda0c53e86dcf435c2 : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53405ef5880542da94ae972d8f9b7bbe : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7161e45b9e6c4d469339a2c71f17dfe9 : SUCCESS in 14m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9dfa7606926e41db8c948d8a5624cf5b : SUCCESS in 30m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/10072780ad294979a6c70dfa0cb4dd27 : SUCCESS in 20m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8e3a5b6be85642ed90d88b291200a06d : SUCCESS in 14m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1801a566ceb4044a853529fe3058c9d : SUCCESS in 21m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7bf44153bab94c88b038974d1c2bc7cb : SUCCESS in 5m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ee7c809b0ec04f209a9ac96c77239c9a : SUCCESS in 7m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a02077a2d1da4244b94e5bad3a2c9509 : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eb867151bc204d30b9fb9c7aa28ba964 : SUCCESS in 14m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3ca5592b431a4ce2a64079b5e2ef5b16 : SUCCESS in 9m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6546f9cd7e7d4de18a677567aa511145 : SUCCESS in 1h 00m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/59f92aec333f4777833a0d4c6c80e6a2 : SUCCESS in 1h 09m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82d7fcd4c7d947e1b934a307fb2a5bf2 : SUCCESS in 1h 32m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/86959b568d6a4a6e884d04dd287c2cc4 : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f12fed65dc5ca207b24df85f6e91d1c749fbba06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 01:27:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b46bf4aa67d09bda9469d04d30f71db606cd8f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 04:42:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"93c2b9c93c1262f79368008a490283f8d479d539","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 04:42:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c36c2014a3645d685f167648fd03c7e : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908265dfb9504a75b5fda4bb36365bff : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c4d7f6dc7b842f9a66d8bcdcbbf29b0 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55db24594b834ae8a0d0f877aeb45adc : SUCCESS in 7m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f8a3f2dda184cb3850820ee0241fd6c : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f90bce10d47c4ebe8a76bc7fb1d4170b : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b7d309bf1b4947d7a72b39d21b896cc5 : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/943ce3b6661544528565caafe0bff7ec : SUCCESS in 11m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3e30ce60ff774449b64457e684389fc0 : SUCCESS in 6m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/80e9f78afa1d4350a5b2187fb08144aa : SUCCESS in 9m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc1bd725c00d454484980fcac00ec95a : SUCCESS in 10m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/64870a1fdbbe488bad7614235f8961aa : SUCCESS in 13m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/21eae4f359d949729fc6149e7aa0ecab : SUCCESS in 22m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7147c6cbd91148b78ae9e8291119e789 : SUCCESS in 14m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5f495cf4a52a4d339b9c35f9baf5e9a1 : SUCCESS in 16m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19af6c346d334971bcc01dd2ba504b64 : SUCCESS in 19m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cb1747a31be049eb915b816eed555c14 : SUCCESS in 5m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/12cba03364c3428c816b6225de62ccb7 : SUCCESS in 7m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/beb5781df1144a06992823ea65992850 : SUCCESS in 8m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ab3d1cdb605345f59020ede753e68911 : SUCCESS in 13m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6661d5e1ec1f46eaaffa746429faf190 : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14f89c6d78b848a09be8175e192a1cd6 : SUCCESS in 55m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c1cb2b23d4b430293bd8ed7eb5ccae9 : SUCCESS in 1h 09m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85ae69152dd64b9a8860b1976b2f4565 : SUCCESS in 1h 18m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c6b01e8692bc4a4e88e9a9cdec1eff95 : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9f1be8adf5cb3d77066e856a766ebe8d5226887d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 04:44:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/818a10eaf29e476a9328cca0999f78bb : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1cec02b450e7f5058b0986223808e0adeab72359","revisions":{"1cec02b450e7f5058b0986223808e0adeab72359":{"kind":"REWORK","_number":1,"created":"2020-02-27 16:19:18.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/09/710309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/710309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/710309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/710309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/710309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/710309/1"}}},"commit":{"parents":[{"commit":"860ebba65255df05cd55501da7d443c45193a3f5","subject":"Merge \"update constraint for oslo.rootwrap to new release 6.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/860ebba65255df05cd55501da7d443c45193a3f5"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-27 16:16:55.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-27 16:16:55.000000000","tz":-360},"subject":"Manual run of generate-constraints","message":"Manual run of generate-constraints\n\nhold back docutils because boto likes to cap things\nhold back msgpack because salt likes to cap things\nhold back alembic because it breaks nova tests\n  - review.opendev.org/708707\n\nChange-Id: I26dd1ca1b06b487248276b11a68f9c6c1067070e\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cec02b450e7f5058b0986223808e0adeab72359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cec02b450e7f5058b0986223808e0adeab72359"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
