)]}'
{"id":"openstack%2Frequirements~710345","triplet_id":"openstack%2Frequirements~master~I5a1a53671b0f617fe2e0b1abe0914a1193f2f8dd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5a1a53671b0f617fe2e0b1abe0914a1193f2f8dd","subject":"update constraint for oslo.utils to new release 4.1.0","status":"MERGED","created":"2020-02-27 22:40:30.000000000","updated":"2020-02-28 09:13:21.000000000","submitted":"2020-02-28 09:11:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710345-1582881102407-97979107","meta_rev_id":"2d01cb0e747c62841e7b664f20b9922e0c08db65","_number":710345,"virtual_id_number":710345,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 09:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-28 07:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-28 07:22:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 22:48:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-28 07:22:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-28 09:11:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a6558c3a2088ed6bc18643ba58817914309fc17","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-27 22:40:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb7958553fe8aa636179e39cbde2a22455014c43","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 22:48:50.000000000","message":"Patch Set 1: Code-Review+2\n\nYey, the bot got smarterer.","accounts_in_message":[],"_revision_number":1},{"id":"d398cea31d04b6729f75e4f95432bade9932fdaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 03:10:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eadcf8f05f744fc28b069fa8514a7b6a : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b6c8d444164cd3a9e1a40394ad8442 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72c73f6b32e4482096df3ce6817c6194 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2dc113a5154c40819a7014c73eb815a4 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a52b44cb0f48adbe50ed08c898c70b : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/685c020d145d427f85912bb12df2e36a : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07bac7873da6435585741ea165cf7311 : SUCCESS in 5m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8940921ff9d94e2d8295aec99484163b : SUCCESS in 10m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e84ce12de53f44539b4c1c69409eb5fe : SUCCESS in 11m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/048044f8539d45abb8834bb63a9acd7a : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/879bb9f932f94bd4b4b0349661b51328 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94179fe35c554b48b2934d9821ef75aa : SUCCESS in 9m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/845422f3eccd44128796412a4a4f1d04 : SUCCESS in 16m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6f80b39f92de4315bde06f57cdcb0e33 : SUCCESS in 25m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1e2c7615f0094a24b477cc19169c8baa : SUCCESS in 16m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b90fdf27d9f54492a452c244dba507f0 : SUCCESS in 16m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7c242eef8244ba182b70094eb89e0cc : SUCCESS in 23m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1edb2122dc4e4dddbcf664b0a06619b7 : SUCCESS in 6m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cc5f92a00ba04335b46eeac887ea3894 : SUCCESS in 8m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd9010e1e7034a0da02a8eafe1f3d777 : SUCCESS in 8m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/545bf8af33d24f2c9dcc62e435a3d07b : SUCCESS in 12m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c07e4ebc159e462b8f960a8ecdda01ca : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/824214fec25343ae866364830dbb7da7 : SUCCESS in 1h 10m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/560f1e92a9b04cdbb704123efc675f57 : SUCCESS in 1h 05m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b43e2140b1d647a8becd4d0acf112e60 : SUCCESS in 1h 32m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7e5a00b3b5604f34a3401352a6ac3309 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":1},{"id":"744d33ed2001bc12beadaf13ccbc9caf0b12392d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-28 03:19:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f134d45cc14666e31efd58824a59338c866d484","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 05:02:46.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"24c874061bfbec6ddd08b32d68df2bfc048ad013","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-28 05:06:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"139a949101c48473c4e232b8a206ad7433adc10d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 06:28:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528db00a073b40da8a28d5533b24d437 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02767c4b564437ea0ec43ce77c5f517 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9309fa40e13e4e89a79156fb479b4301 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce259fa5d21c428085b31fd9e7290e48 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13499a2abaed4a86a571e8424ee8b7c3 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0aae0fe89c434e2683850618f3503731 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/194a2dadf6fc4dfdb68dabe3fe978d0f : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c0ce7558cd814485bba4cd28154d2237 : SUCCESS in 8m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/82d714110d3f4515aef66ab862d118c5 : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/02422308cdb74e0ba6d0741b3d7aaef1 : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a761dfe9af7b487d8a2828c1b4bead69 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/856fa6f6c7934687bc9b9e101aff4b72 : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6ee7f1b65ddc49158ec0bbee4d419266 : SUCCESS in 13m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/19bdcdab53494a13a61796452cd38bbc : SUCCESS in 24m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c65fa0f0c7434c30bbd2c0aab933798e : SUCCESS in 13m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/700eb4711358491b9923a74a50bc5768 : SUCCESS in 14m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bedbfff953f74be3b3b9c1588946a6c3 : SUCCESS in 19m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/258257b901544f9d9c57a6d13020d60e : SUCCESS in 5m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0d03ae4184a140369ba6470383c484f8 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6691e98174b84b7b84389614ba50bf29 : SUCCESS in 7m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c8130e96b18143fcb59b578bf88e28c3 : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2c874e23ea4240a7b154b822ab70906c : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e5b153bcac742e0b87328df4ce42a8a : SUCCESS in 59m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7967b7d31daf44dabd7280a405195d7a : SUCCESS in 1h 09m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14e70741994d4d1d8a35cf91b8c06eb5 : SUCCESS in 1h 20m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bd0b460606a2466980d8492385ccb849 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":2},{"id":"428217f7eb05c3a42d45088a4716819a4bedcae1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-28 07:22:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4918c916881d2cf57e299b92f0862fe373574966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 07:22:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1991f72289605cf5b73be21c644986d62d527b86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 08:59:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca10944ee92842828c1a0a1cda850e8f : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d759c63daa49058fb565680941ebba : SUCCESS in 9m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa0947027b394cc1a64b0f63d7770e0f : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed68665cf2174d96b49a6aaa5bc14fea : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad4807389fc3436ab647e920d49a6fdf : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51328f8aa6f04c2a843985612b93162f : SUCCESS in 10m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1dc3cdd26fac4008b2cf65fcdc6ee19e : SUCCESS in 9m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6a6918c114ac4622b952e69b6f32ac98 : SUCCESS in 10m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ceb2f06f16e745308ebb6c66d19d3267 : SUCCESS in 11m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/103fc93a1555437c9d8d3a08541d9621 : SUCCESS in 9m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/885fe1194b344ad8a05a7e84db912abe : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/960311e25c624b789b495a7186772a88 : SUCCESS in 9m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/84c98194f9784dc190f5077e820bf8db : SUCCESS in 26m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ca607b396f87427d9348d07a2feebac1 : SUCCESS in 30m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/61144776d0db4dbc9fa5b50e92885c7c : SUCCESS in 14m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bcd608d7e1d04c31ab4d131dfa67c6de : SUCCESS in 23m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/936cab02d75c45ecb1e33e19634d160f : SUCCESS in 17m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0efcfbb849314bfabf2ecc808068674b : SUCCESS in 5m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/18c3b3a29f9d4f73a967ac89bb610bf4 : SUCCESS in 8m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24efc79fc8a44396adab5fb913778ae5 : SUCCESS in 9m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4d9aa5a47cb846c7aa205a7ca5b361a1 : SUCCESS in 12m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6a1b4afb09204e1fa2b6a4e79290490d : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/672253e42abd45f5ae2f7460e9bf12a1 : SUCCESS in 1h 09m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4cdeb73c96a44cbd888fe744e82091b9 : SUCCESS in 1h 19m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/baf284fc7bf14b01baee5a63d34cbe72 : SUCCESS in 1h 16m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/10a4c6493ce543e98986c861e9e7faeb : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1b980fb09ed047fbd21ac210be7e37b63ec267a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 09:11:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf07ede33174bb786c5fe85405bd7e1 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b077ea71a8549aa840234fd81933be2 : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbd0386b12fc4a39ae5a823fad47f4f7 : SUCCESS in 8m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb172ee602c04fd59979207679d5f65b : SUCCESS in 9m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72b8a7a93faf4840881fb5caa8afe15a : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/424424a9fcf042c19777fedda0abef9e : SUCCESS in 9m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f90fbfbbb248414d859f0cc2cbcf7ce4 : SUCCESS in 14m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e6572c72edd1450ea68d290dc9bca975 : SUCCESS in 13m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3cb2951c15f8468eab26541f67d817ee : SUCCESS in 7m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7e055ec346d34c34a133fe37ea41da44 : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c66d61d80cb64ff8a9b0f832148d73e9 : SUCCESS in 11m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/479f80a71e1e443ea2d43c5a0f9651fd : SUCCESS in 14m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/075f54c60ab147bfbf143f5b94172f5c : SUCCESS in 26m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/76df310346834bf0a8d4d0b51e5302d0 : SUCCESS in 14m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f45279850f7243a78791cb319a62d164 : SUCCESS in 15m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/998829e2e1d949f59d987b4541073dae : SUCCESS in 18m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/01101e0a66b04ab8a63807dfdedc7d2f : SUCCESS in 7m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bc5a5b1350204b8dbc06e8c8cf9f3a18 : SUCCESS in 10m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d43dceff5994951bea035cacef2a86c : SUCCESS in 10m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/817833bd84cf49d38b87be81cba43339 : SUCCESS in 12m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c16df7312e3e4ddc94e02d5a676f609c : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5c8fed48651409b954f2aaf55213d99 : SUCCESS in 54m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ffa3e451e7d4f9997be4a57a49dfc51 : SUCCESS in 1h 10m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fec30c82cb95462a916b9e296936d579 : SUCCESS in 1h 31m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/21931cc061474b0f9ee36660282f1832 : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":2},{"id":"01392f71ab9e3079d6172d37661d723f5487f0b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 09:11:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"15d044f3c803488b82423c70891ac50a1d4db8e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 09:13:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3ae1bde6644606aebf566d813b8cde : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4d0ca1128fe11cd72154c6340e5751f099ff3bd","revisions":{"498ddd0f570c39e695a95fa777c3112e5e0b8033":{"kind":"REWORK","_number":1,"created":"2020-02-27 22:40:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/710345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/710345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/710345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/710345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/710345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/710345/1"}}},"commit":{"parents":[{"commit":"860ebba65255df05cd55501da7d443c45193a3f5","subject":"Merge \"update constraint for oslo.rootwrap to new release 6.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/860ebba65255df05cd55501da7d443c45193a3f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 22:40:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 22:40:27.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 4.1.0","message":"update constraint for oslo.utils to new release 4.1.0\n\nChange-Id: I5a1a53671b0f617fe2e0b1abe0914a1193f2f8dd\nmeta:version: 4.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I976720f7bf9471b9017466a5a40956f9112485fc\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/498ddd0f570c39e695a95fa777c3112e5e0b8033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/498ddd0f570c39e695a95fa777c3112e5e0b8033"}]},"branch":"refs/heads/master"},"e4d0ca1128fe11cd72154c6340e5751f099ff3bd":{"kind":"REWORK","_number":2,"created":"2020-02-28 05:06:16.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/45/710345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/710345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/710345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/710345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/710345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/710345/2"}}},"commit":{"parents":[{"commit":"ee9b4286ffd1fc527371ceb2ce20f6e7aa62a323","subject":"Merge \"update constraint for os-win to new release 5.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee9b4286ffd1fc527371ceb2ce20f6e7aa62a323"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-27 22:40:27.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-28 05:06:06.000000000","tz":-360},"subject":"update constraint for oslo.utils to new release 4.1.0","message":"update constraint for oslo.utils to new release 4.1.0\n\nChange-Id: I5a1a53671b0f617fe2e0b1abe0914a1193f2f8dd\nmeta:version: 4.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I976720f7bf9471b9017466a5a40956f9112485fc\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4d0ca1128fe11cd72154c6340e5751f099ff3bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4d0ca1128fe11cd72154c6340e5751f099ff3bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
