)]}'
{"id":"openstack%2Frequirements~710393","triplet_id":"openstack%2Frequirements~master~Ide7313ed50b552b982307b76574d92fb8517683b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ide7313ed50b552b982307b76574d92fb8517683b","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-02-28 06:31:28.000000000","updated":"2020-02-28 19:10:17.000000000","submitted":"2020-02-28 19:07:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710393-1582916866873-577aa814","meta_rev_id":"a4ff50dccdf152203904382d87538e121968a0de","_number":710393,"virtual_id_number":710393,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 19:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-28 15:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-28 15:46:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 15:46:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-28 19:07:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"367a716e22708e3fd53dd4333ce4fe0a3feef25e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-28 06:31:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5d0c499e455c244af084886feb6b02c91e30b20","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-28 07:25:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f8081622889c8646a09f80b67aa13d69713f4a0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-28 07:45:28.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"d7e265e5bc491dc2baf2bf45da2b5895d242d0bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 08:52:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d92e21ac1a4ac484929ae0f6c5351f : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0197ce2e8b4c4803a5c9c03818054bca : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e50579f2b0047798df28741f5756993 : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd4f22e962e54213a1cd9f297b91460c : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0c2afdf069e4bf084b24bed67432c39 : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/819693e5684743f4b42ee787553cfc33 : SUCCESS in 8m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45ad018ab1b24a69892cefa946a351d9 : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d362a776676544cea0cfd2e3436bed9c : SUCCESS in 10m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8e9c4cabf44b4fabada23d5df6f16318 : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/30a611d4c8e842d69dcb15767c2d0ba8 : SUCCESS in 6m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c152ce9a29b5459b980035f5625c3e54 : SUCCESS in 14m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6c49c3e08f949deac0a4fd95c704d75 : SUCCESS in 16m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/52fc8b19b7f94936bd25eb18b1f22ab1 : SUCCESS in 39m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eef5c526b64541a0859ad0f8b02d18ad : SUCCESS in 24m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/09f9b40fa59949169f763d37cc2379bd : SUCCESS in 17m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/16fbb554c6f64d6192141766b56c1c50 : SUCCESS in 14m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e0dba7a4de44d4fa762040f03a57853 : SUCCESS in 18m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/71f0a66f32494c13b329a248e5d19d2a : SUCCESS in 7m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/90d8caadb0d44d88860547847a794a3a : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1dbc81bacd3c45129ff411d8b43d5741 : SUCCESS in 10m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9b51b21448244e1cb08792b285fa1e05 : SUCCESS in 15m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/86d65be60ef24e89b773111bf2182933 : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/194f27271d754a8bb6b5970ce7778332 : SUCCESS in 1h 01m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9843acf461e4f64b196f5aeca8d233e : SUCCESS in 1h 08m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b50e0d070a624843be1fb07fb8716b80 : SUCCESS in 1h 16m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/31b5f3c81534447b9eb358d9a953e4c0 : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":2},{"id":"16aaec1877f9ac6dff91555815ab140b2f4c216c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-28 15:46:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e10248741e243dc79ab841c2a4d090b5ef45929","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 15:46:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"77352ad3ebb3b55fb02a1df1ee57aca9092af8f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:07:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d95e85510c04671a17323758db1ca9b : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3046432928ad4986bf1bb475272f2b6a : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/562256c3f4904b5bae01cb8093163a2b : SUCCESS in 9m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6653ee04e3f74c10926a85ebe37e0dc7 : SUCCESS in 7m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/394c533b5649449aba9b61ca570c13da : SUCCESS in 7m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78859f7c39644a6e95e216748de8c916 : SUCCESS in 8m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8e7747b9b1504a2d87daa285f7bb7029 : SUCCESS in 9m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/14a804042b924d4883f49e9f9fcb2345 : SUCCESS in 9m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7fa6b164cd014e37b98c575c00255393 : SUCCESS in 6m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f634bb62f2294ff3983e9e1bb153766a : SUCCESS in 14m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17593bba6fae45a8867f3d84048681cb : SUCCESS in 16m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3ce17d1358ec4970aa9206dd419b68c4 : SUCCESS in 18m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a391e646b1d44191b50d3d7236788b65 : SUCCESS in 28m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c85d654380694186b4935bc086c5f9a8 : SUCCESS in 21m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/aae27ebd2a354d46ae4993147c1f9251 : SUCCESS in 16m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3b168902ec54516b906756e92609e4e : SUCCESS in 16m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/991b6e3c439449338ac9b7f056e636ac : SUCCESS in 7m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3b834f27365f4a4e99baba9d9caa273e : SUCCESS in 8m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/429780b8209d4098bd2f6c6f87041ef0 : SUCCESS in 9m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6d7ae2b94fe24ba3a21e5957ce9c898a : SUCCESS in 14m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d22e6fd29e944284ba40f844401bd016 : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36c5c5e6adf14a3bb99d7b76593efeae : SUCCESS in 53m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c464a21d56d410a97dc747a7861bbec : SUCCESS in 1h 33m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45d0f440acc6402c81140d06e55f6126 : SUCCESS in 1h 18m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e48bc622c01d4638a8aca7cc9c55a479 : SUCCESS in 1h 09m 11s","accounts_in_message":[],"_revision_number":2},{"id":"437a0aa6bc194616d589d2420fac1e6cb510e3ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:07:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a4ff50dccdf152203904382d87538e121968a0de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:10:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4f1a6a63cc4948b77e50678e1faa61 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e7f35d36a9f1eabae28768d906e009cd41f68eb","revisions":{"153be772ea9d973f40adb9c858929763d5d0bf49":{"kind":"REWORK","_number":1,"created":"2020-02-28 06:31:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/710393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/710393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/710393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/710393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/710393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/710393/1"}}},"commit":{"parents":[{"commit":"ee9b4286ffd1fc527371ceb2ce20f6e7aa62a323","subject":"Merge \"update constraint for os-win to new release 5.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee9b4286ffd1fc527371ceb2ce20f6e7aa62a323"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-28 06:31:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-28 06:31:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ide7313ed50b552b982307b76574d92fb8517683b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/153be772ea9d973f40adb9c858929763d5d0bf49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/153be772ea9d973f40adb9c858929763d5d0bf49"}]},"branch":"refs/heads/master"},"5e7f35d36a9f1eabae28768d906e009cd41f68eb":{"kind":"REWORK","_number":2,"created":"2020-02-28 07:25:31.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/93/710393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/710393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/710393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/710393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/710393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/710393/2"}}},"commit":{"parents":[{"commit":"ee9b4286ffd1fc527371ceb2ce20f6e7aa62a323","subject":"Merge \"update constraint for os-win to new release 5.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee9b4286ffd1fc527371ceb2ce20f6e7aa62a323"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-28 06:31:27.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-28 07:25:17.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nhold back docutils because boto likes to cap things\nhold back msgpack because salt likes to cap things\nhold back alembic because it breaks nova tests\n  - review.opendev.org/708707\n\nChange-Id: Ide7313ed50b552b982307b76574d92fb8517683b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e7f35d36a9f1eabae28768d906e009cd41f68eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e7f35d36a9f1eabae28768d906e009cd41f68eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
