)]}'
{"id":"openstack%2Frequirements~710555","triplet_id":"openstack%2Frequirements~master~If65e7aff4a90e2c8ebd46cc43a21b2860a35b909","project":"openstack/requirements","branch":"master","topic":"bp/tosca-csar-mgmt-driver","hashtags":[],"change_id":"If65e7aff4a90e2c8ebd46cc43a21b2860a35b909","subject":"Added sqlalchemy-filters requirement","status":"MERGED","created":"2020-02-29 11:38:46.000000000","updated":"2020-03-18 23:35:42.000000000","submitted":"2020-03-18 23:34:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710555-1584574464169-46c24852","meta_rev_id":"aec244ed825c65a328c139de3ca5b08a9059b924","_number":710555,"virtual_id_number":710555,"owner":{"_account_id":31580,"name":"Sameer Thakur","email":"sameer.thakur@nttdata.com","username":"sameer.thakur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-18 23:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-14 16:16:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},{"value":2,"date":"2020-03-18 21:12:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-14 18:02:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2020-03-17 08:18:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},{"value":1,"date":"2020-03-18 21:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-10 04:52:33.000000000","updated_by":{"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},"reviewer":{"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},"state":"REVIEWER"},{"updated":"2020-03-14 16:16:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-14 18:02:05.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-03-17 08:18:47.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2020-03-18 21:12:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-18 23:34:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8edf37cccf9569a0fcc92f994457043ad537db89","author":{"_account_id":31580,"name":"Sameer Thakur","email":"sameer.thakur@nttdata.com","username":"sameer.thakur"},"date":"2020-02-29 11:38:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0931a6cbbce674d1f53e36c4464e7123f421e12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 13:19:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1875c4dd680f4cb7b48026143a609606 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7842f68b89142ae888700cee9892b2f : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3256ffbbcd948d5a723bca71f4a31ba : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7ccc3c97e344253bc8610399273db22 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f4dff1d2c2d493ca784bf1c44671a42 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e715512e99784ac5aff07b88fbc2db50 : FAILURE in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e57a190ff1847cd8923741982960fa1 : SUCCESS in 52m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f5bdaf423ac4d278ea1ce4b790d3776 : SUCCESS in 1h 08m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10857513c3e449148fc1ce93b25a0bb6 : SUCCESS in 1h 38m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4dcd8afd52694b4abcb127fcc43ec086 : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"de0d4ea879f318e3c24b081e87bbc9cccc980b5b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-01 18:45:42.000000000","message":"Patch Set 1: Code-Review-1\n\nplease add a version to upper-constraints as well as answer the questions from https://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements in the commit message","accounts_in_message":[],"_revision_number":1},{"id":"759e030c977b8cf92cd444ad5eae59f740ca5e82","author":{"_account_id":31580,"name":"Sameer Thakur","email":"sameer.thakur@nttdata.com","username":"sameer.thakur"},"date":"2020-03-06 09:12:56.000000000","message":"Topic changed from Attribute_Filters_Req to bp/tosca-csar-mgmt-driver","accounts_in_message":[],"_revision_number":1},{"id":"c63fd7559563699a8b3cf8c66a17b64a3ca36602","author":{"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},"date":"2020-03-10 04:52:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"253f935a80cd55cb1c4a331092d6842abcbf95c9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-10 04:58:34.000000000","message":"Patch Set 2:\n\nsome PRs seem stale","accounts_in_message":[],"_revision_number":2},{"id":"37cdde147a87c3279db546162dccb2d376f81bb6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-10 05:04:52.000000000","message":"Patch Set 2:\n\nseems like simple code though","accounts_in_message":[],"_revision_number":2},{"id":"825446bc9f07d24d321a6cc8221a33d41036584c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 06:21:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f34c377606746bfbe6862b49e9c6bf9 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/924d6bac452f4d219d851d5b939a41ff : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/288c845e43b447e1921093351200947d : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d86b8c070d654ce689ff3fa0dc73896a : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4b90cfa41b9419fb1097cc275e432c3 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5eedd9bbc41b4c008bdf673588ab8b04 : FAILURE in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af052ba616fa489394fd72374865e73d : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a039316ec6cc439093e9739bd5f427d6 : FAILURE in 4m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/78784779276045cd97759e2f74eb23f3 : SUCCESS in 9m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9cab9543a3ca40bbb04a053c9e1d2873 : SUCCESS in 4m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/804cc975473c4798bea7ca5bf054c85b : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b830fdf8b3a046efb6f1217950429f53 : SUCCESS in 6m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7326524461544a39801b3972e4f80001 : SUCCESS in 16m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f73b1d4f549b4965a710a5d3dff434e8 : SUCCESS in 21m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/55f5ee3d409b457eaffebcb9e096b29d : SUCCESS in 12m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cb34c1097b014c5ab8bed256d3ee516d : SUCCESS in 15m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84d3e191c81640748ceb08b82432ad26 : SUCCESS in 15m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/abf83a64d7bf43628f9ec28023d34170 : SUCCESS in 4m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ee626cbc6fe14ecaac16b6cc1f7b2fd4 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b44c9e191e2646388393a2ef11354067 : SUCCESS in 6m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0d0830dfd24e4a0f908222fe1e804c68 : SUCCESS in 14m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f82c1974b4f44482853f8a82c94177ed : SUCCESS in 9m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec4a15cf16024874981db7739caeb7df : FAILURE in 55m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e068cf148c940a6bd443968e326ab67 : SUCCESS in 1h 05m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3574c972442a4b59b73b1aa58b7f3319 : SUCCESS in 1h 24m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7158a95313e04ccaa40b10c3e0740bf7 : FAILURE in 1h 02m 55s","accounts_in_message":[],"_revision_number":2},{"id":"45fed295b3a4ddbe8fd5ddcf26e54ffeafb7ed6f","author":{"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},"date":"2020-03-10 07:38:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f8b48e5b8915047a2ed512fd3df31c7cb4ac8801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 09:17:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93762507e69e40cba6657dd0f13f6596 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef9765d1a0a4490905282964fc261fa : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7de7097938994ae89992a228c5d40891 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cbea688d68c4577a87290c789b67bb7 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/316d21598b954a1aa550161d21dfa87d : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c553b3eda32d4cff88fd3928266dbbf9 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61bf3754844743b091a4f676e89248c3 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/46ae91758ceb49c38ba7ea8db3957502 : SUCCESS in 9m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0c0e62465a6c4b75a1457a0a3a399987 : SUCCESS in 10m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7f4e376f4d2d40d29d0e25f6f55e2a1d : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2435328cec924d209aceae66b2d3a196 : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cbfaeae6ac1b4ef6bd2a926703af7c53 : SUCCESS in 6m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3bb14da788ad459aba88288d94f4d1fb : SUCCESS in 20m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/185ed7d0d083462b9806809dba1d2872 : SUCCESS in 26m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/62866deb7ea04576b6309986ac27298b : SUCCESS in 14m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1346b16d79354929bd48edbf4f02d69e : SUCCESS in 17m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47ec33c1877f4adabd712d76c8566d68 : SUCCESS in 23m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/90b8c74ad3fe4556a175116a41a9297f : SUCCESS in 6m 39s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bdf892649c5a42c5aef24d31968e5d89 : SUCCESS in 9m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c774d797db14d49a2e5188e8bb59893 : SUCCESS in 7m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c3f56b19e1254063838d456b99e8d10a : SUCCESS in 20m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a227ea88c5d1408abf83e1237dd33ebd : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf57fcdf41f445d487645d1f26b6c196 : FAILURE in 1h 10m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b53c779a208c416fb1a67043cab487a7 : SUCCESS in 1h 24m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/825edab11d624be59a029a01e3339507 : SUCCESS in 1h 36m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b6aca302c7c64732a2ceed97b34b7bdb : FAILURE in 1h 19m 23s","accounts_in_message":[],"_revision_number":3},{"id":"a91a985926ed1773e4eeb5ee6b470ad021fcea8b","author":{"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},"date":"2020-03-13 01:01:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"44621443bfab48e2da7a058daf54564346f99d30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 02:38:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6aca0d1a17e4cf980870cb3548ab9a8 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc23f5c262c14aae9cdfae09a5915988 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d8c27b8befb41af9b541b133a54b739 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e790ac83c9ee4888815ba8aaeecf41c2 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e764fc33b27c45f084c063d82d563125 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/261a9ef22b10472c84f28f5a4bd5b912 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff943b9ea20c4a1e8431075855645671 : SUCCESS in 3m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/16e4b8d878714d51a35ef1688bbb67a3 : SUCCESS in 8m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0680c6947fee4a34915f3c19ea8842d9 : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c0dd0e9091bf43d38c0d690663d191e0 : SUCCESS in 4m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd00fc584ef7437b8fd5e34c00196bba : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/700e3bef71244aa687c640ec0121cf2e : SUCCESS in 6m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a42aa05882b94f70a93c25cefdc60e1e : SUCCESS in 12m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/62c17f8af488419784819547a282259d : SUCCESS in 28m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/221233b3406147b69a70ad326363b45b : SUCCESS in 12m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ed339a0d780b40e2bbba71013f88f4d6 : SUCCESS in 12m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e9ed2d82551494da5705749be92c632 : SUCCESS in 15m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1409a2a183154ec3bcd2395ad09d56be : SUCCESS in 4m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6f284cd2dfb9465894067d5c9a5a238e : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/703960571f74494db5c7af87c52ec88e : SUCCESS in 6m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/72203cacd3be4f1c85ac810da24afce7 : SUCCESS in 8m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a5c9c1f7d0d34abcbfed5a0dc10e381a : SUCCESS in 9m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05ff0c68065d46d9beda47b38ddd8403 : SUCCESS in 53m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/063995a6104547dc93d95f7d2abcb762 : SUCCESS in 1h 04m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66dadef524454377ab88da24f9dbbff2 : SUCCESS in 1h 34m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/da55e8bd44f143dc8d254efa51a3c2f1 : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":4},{"id":"8b80249784e11e35c1b3f66cf9ddb001117492f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-14 16:16:12.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0a57977289f5e160f2938c088afd65ca2a868beb","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-03-14 18:02:05.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"71eda790c5a496ca252f090dd03a03bc212207cd","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-03-17 08:18:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f8250ff941a881fa1979bbbe588322e7e45a8ec5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-18 21:12:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLooks fine to me.","accounts_in_message":[],"_revision_number":4},{"id":"7dbf580e9d97bcd4182bba61b3a956f35533cf11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 21:12:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"814ec247c07ed65cbc2e301f3b2032f844a46fa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 23:34:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf8b07e4df9d4574a5f640331106339b : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ffc506251b444fbac9a231337026191 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9068660aa471416fbfc1d9ef690462e6 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cf95911b9c343d59dd8ff508d6c49c1 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d90566197994751b046cb7ab8da33f7 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/330b90a712774279a971a3236fda9a16 : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3614c4d59e194ad79efb3343322b789e : SUCCESS in 7m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/960c1c070d7046e1b6fb00c978d6329a : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/055443a8b3814dd7961e19c5bd36ae4a : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6a19e3ecfae9455d862d22a4f5835e7b : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d0aa8700a864c48892412de3797e68c : SUCCESS in 9m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/760e9efa99744ab5ac35abde17a3ad4c : SUCCESS in 12m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d41006222e624b04ad760b8c08fd791e : SUCCESS in 25m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3136ebb16e754f9999f685605ab3e73d : SUCCESS in 12m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7f4823671a414d078b17ce6e059f7de4 : SUCCESS in 12m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08e84d5afc8645258817ad002e879661 : SUCCESS in 14m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6c4092dfd8124f06883742ccf46fb63e : SUCCESS in 5m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/12660945af08445ca82b2f464aacad3a : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bad74ab2367d4a4e83e943942a801b5f : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f34c0929037941beb1a0beeabc3c1181 : SUCCESS in 8m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cd14164f4c7945729664ee90aa35b840 : SUCCESS in 9m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c727a772396742558757a5cd7f735840 : SUCCESS in 59m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69038943bd434bbfb01e486b65f7954e : SUCCESS in 1h 14m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c960b7408ecb48b4979e3aaf3d1d05a6 : SUCCESS in 1h 22m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2826fe13dcab478ca6b4a2a5a6e7e7b4 : SUCCESS in 50m 23s","accounts_in_message":[],"_revision_number":4},{"id":"4932a9c4ce7fb3010c3963bb1f4b354040ceac42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 23:34:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"aec244ed825c65a328c139de3ca5b08a9059b924","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 23:35:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da717e89a3c84e35bdb00985c69ce24e : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"72d075a74ce3c6230c7d4774a49ae94e18a07a90","revisions":{"c9ab549c294a5be61ed2747fb7f030ed03497bcd":{"kind":"REWORK","_number":1,"created":"2020-02-29 11:38:46.000000000","uploader":{"_account_id":31580,"name":"Sameer Thakur","email":"sameer.thakur@nttdata.com","username":"sameer.thakur"},"ref":"refs/changes/55/710555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/710555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/710555/1"}}},"commit":{"parents":[{"commit":"7a04f4e96205f16e74a9a08fe60b168c187516d6","subject":"Merge \"Fix missing \u003d in neutron-lib marker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a04f4e96205f16e74a9a08fe60b168c187516d6"}]}],"author":{"name":"sameer.thakur","email":"sameer.thakur@nttdata.com","date":"2020-02-17 09:26:24.000000000","tz":330},"committer":{"name":"sameer.thakur","email":"sameer.thakur@nttdata.com","date":"2020-02-17 09:36:30.000000000","tz":330},"subject":"Added sqlalchemy-filters to global requirements","message":"Added sqlalchemy-filters to global requirements\n\nThis module used in the implementation of Attribute Filtering of\nVNF Packages defined in NFV_SOL005 section 9.4.2.3.2. This module\nis used to add filter criterion to SQL alchemy queries. These\ncriterion were based on attribute values of VNFPackages\npassed in the GET request to list VNFPackages.\n\nChange-Id: If65e7aff4a90e2c8ebd46cc43a21b2860a35b909\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9ab549c294a5be61ed2747fb7f030ed03497bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9ab549c294a5be61ed2747fb7f030ed03497bcd"}]},"branch":"refs/heads/master"},"ccd2a89359e2d1b9d2b882f302476457fbe95a83":{"kind":"REWORK","_number":2,"created":"2020-03-10 04:52:33.000000000","uploader":{"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},"ref":"refs/changes/55/710555/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/710555/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/710555/2"}}},"commit":{"parents":[{"commit":"2f2c92a3a3bc53eb945d42166c11ded359f5db33","subject":"Merge \"Updated from generate-constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2c92a3a3bc53eb945d42166c11ded359f5db33"}]}],"author":{"name":"sameer.thakur","email":"sameer.thakur@nttdata.com","date":"2020-03-10 04:49:23.000000000","tz":0},"committer":{"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","date":"2020-03-10 04:51:19.000000000","tz":0},"subject":"Added sqlalchemy-filters requirements","message":"Added sqlalchemy-filters requirements\n\nhttps://pypi.org/project/sqlalchemy-filters/\nUseful for Filter, sort and paginate SQLAlchemy query objects.\n\n1. Is the library actively maintained?\n   Yes. The library had 2 releases in 2019.\n\n2. Is the library good code?\n   Yes. It also has tox targets for testing.\n\n3. Is the library python 3 compatible?\n   Yes.\n\n4. Is the library license compatible?\n   Yes. Apache-2.0\n\n5. Is the library already packaged in the distros we\n   target (Ubuntu latest / Fedora latest)?\n   No.\n\n6. Is the function of this library already covered by other\n   libraries in global-requirements.txt?\n   No.\n\n7. Is the library required for OpenStack project or related dev\n   or infrastructure setup?\n\n   This library is used in Tacker project for the implementation of\n   Attribute Filtering of VNF Packages defined in NFV_SOL005 section\n   9.4.2.3.2. This module is used to add filter criterion to SQL\n   alchemy queries. These criterion were based on attribute values of\n   VNFPackages passed in the GET request to list VNFPackages.\n\n8. If the library release is managed by the Openstack release\n   process does it use the cycle-with-intermediary release type?\n   Not aware of it.\n\n9. Do I need to update anything else?\n   No.\n\nChange-Id: If65e7aff4a90e2c8ebd46cc43a21b2860a35b909\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccd2a89359e2d1b9d2b882f302476457fbe95a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccd2a89359e2d1b9d2b882f302476457fbe95a83"}]},"branch":"refs/heads/master"},"3e95673d051dfcde4354344b617b6f76b55cc502":{"kind":"REWORK","_number":3,"created":"2020-03-10 07:38:34.000000000","uploader":{"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},"ref":"refs/changes/55/710555/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/710555/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/710555/3"}}},"commit":{"parents":[{"commit":"2f2c92a3a3bc53eb945d42166c11ded359f5db33","subject":"Merge \"Updated from generate-constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2c92a3a3bc53eb945d42166c11ded359f5db33"}]}],"author":{"name":"sameer.thakur","email":"sameer.thakur@nttdata.com","date":"2020-03-10 04:49:23.000000000","tz":0},"committer":{"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","date":"2020-03-10 07:28:55.000000000","tz":0},"subject":"Added sqlalchemy-filters requirements","message":"Added sqlalchemy-filters requirements\n\nhttps://pypi.org/project/sqlalchemy-filters/\nUseful for Filter, sort and paginate SQLAlchemy query objects.\n\n1. Is the library actively maintained?\n   Yes. The library had 2 releases in 2019.\n\n2. Is the library good code?\n   Yes. It also has tox targets for testing.\n\n3. Is the library python 3 compatible?\n   Yes.\n\n4. Is the library license compatible?\n   Yes. Apache-2.0\n\n5. Is the library already packaged in the distros we\n   target (Ubuntu latest / Fedora latest)?\n   No.\n\n6. Is the function of this library already covered by other\n   libraries in global-requirements.txt?\n   No.\n\n7. Is the library required for OpenStack project or related dev\n   or infrastructure setup?\n\n   This library is used in Tacker project for the implementation of\n   Attribute Filtering of VNF Packages defined in NFV_SOL005 section\n   9.4.2.3.2. This module is used to add filter criterion to SQL\n   alchemy queries. These criterion were based on attribute values of\n   VNFPackages passed in the GET request to list VNFPackages.\n\n8. If the library release is managed by the Openstack release\n   process does it use the cycle-with-intermediary release type?\n   Not aware of it.\n\n9. Do I need to update anything else?\n   No.\n\nChange-Id: If65e7aff4a90e2c8ebd46cc43a21b2860a35b909\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e95673d051dfcde4354344b617b6f76b55cc502"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e95673d051dfcde4354344b617b6f76b55cc502"}]},"branch":"refs/heads/master"},"72d075a74ce3c6230c7d4774a49ae94e18a07a90":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2020-03-13 01:01:52.000000000","uploader":{"_account_id":31648,"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","username":"pbhole"},"ref":"refs/changes/55/710555/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/710555/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/710555/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/710555/4"}}},"commit":{"parents":[{"commit":"fcf3d1c7732848c5389ae33c296763cd52213dbc","subject":"Merge \"update constraint for ironic-lib to new release 4.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcf3d1c7732848c5389ae33c296763cd52213dbc"}]}],"author":{"name":"sameer.thakur","email":"sameer.thakur@nttdata.com","date":"2020-03-10 04:49:23.000000000","tz":0},"committer":{"name":"Prashant Bhole","email":"prashant.bhole@nttdata.com","date":"2020-03-13 01:01:17.000000000","tz":0},"subject":"Added sqlalchemy-filters requirement","message":"Added sqlalchemy-filters requirement\n\nhttps://pypi.org/project/sqlalchemy-filters/\nUseful for Filter, sort and paginate SQLAlchemy query objects.\n\n1. Is the library actively maintained?\n   Yes. The library had 2 releases in 2019.\n   Since there was a concern whether it is being actively maintained,\n   we contacted with the maintainer to ask the status. Below link to\n   the issue confirms that project is actively maintained, pending\n   PRs will be handled soon.\n   https://github.com/juliotrigo/sqlalchemy-filters/issues/42\n\n2. Is the library good code?\n   Yes. It also has tox targets for testing.\n\n3. Is the library python 3 compatible?\n   Yes.\n\n4. Is the library license compatible?\n   Yes. Apache-2.0\n\n5. Is the library already packaged in the distros we\n   target (Ubuntu latest / Fedora latest)?\n   No.\n\n6. Is the function of this library already covered by other\n   libraries in global-requirements.txt?\n   No.\n\n7. Is the library required for OpenStack project or related dev\n   or infrastructure setup?\n\n   This library is used in Tacker project for the implementation of\n   Attribute Filtering of VNF Packages defined in NFV_SOL005 section\n   9.4.2.3.2. This module is used to add filter criterion to SQL\n   alchemy queries. These criterion were based on attribute values of\n   VNFPackages passed in the GET request to list VNFPackages.\n\n8. If the library release is managed by the Openstack release\n   process does it use the cycle-with-intermediary release type?\n   Not aware of it.\n\n9. Do I need to update anything else?\n   No.\n\nChange-Id: If65e7aff4a90e2c8ebd46cc43a21b2860a35b909\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72d075a74ce3c6230c7d4774a49ae94e18a07a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72d075a74ce3c6230c7d4774a49ae94e18a07a90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
