)]}'
{"id":"openstack%2Frequirements~711027","triplet_id":"openstack%2Frequirements~master~I0b0191e57d9dd4fe144128e1c683d307098fe144","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0b0191e57d9dd4fe144128e1c683d307098fe144","subject":"update constraint for oslo.cache to new release 2.1.0","status":"MERGED","created":"2020-03-03 14:33:57.000000000","updated":"2020-03-04 04:10:10.000000000","submitted":"2020-03-04 04:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711027-1583294916355-cca37f66","meta_rev_id":"a376f8c5d730f38b5074abd8cedd43716615157d","_number":711027,"virtual_id_number":711027,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 04:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-04 02:31:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-04 02:31:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 02:31:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-04 04:08:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b137de2e9025eea05bc2576bd0f192b3ea6e8306","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-03 14:33:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e9fb405372d0bbb4e6fbf645daad671a2b32895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:01:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024aa6bbc7c44170ae9f89dd10a9ac84 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91304da634f74b489ebaafc32ae6b3ca : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52629cd3636e4e02a8693ad94e0f49bf : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c4c8d243743457baa7288306d0a2b20 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1d6a697492449e7a6b8715fe7ac88fb : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ea4f6a52c3d48f0a7ea5f554934350f : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecc0f7a294d84d20aea094880ed019ab : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ebb8db65d5aa4d74bf689ff48736daf2 : SUCCESS in 11m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7a6715ea8629489993587570982dc506 : SUCCESS in 11m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/db584163d8be4d138031a0f02267df35 : RETRY_LIMIT in 4m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1f6e0fe4f7054c1dbe2bb0915a94953b : RETRY_LIMIT in 3m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/984399e82bad4808a6a6e058b99083c4 : SUCCESS in 8m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8818efb16d3c42eb8c0ab26a99469282 : RETRY_LIMIT in 4m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9999d2a55b984606b2d6378721d632bc : SUCCESS in 35m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4d434e89eff642f893e58a2f1f3f06bc : SUCCESS in 14m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/eb80e0e77fa34a11af54c2027cbf3a18 : SUCCESS in 13m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e93a79de45ad47a98389ad0202055cef : SUCCESS in 16m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/09ee4e595a6d48c9aab367cb2a087f65 : RETRY_LIMIT in 3m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bd6c572db73e46468226cd03a07c5be2 : SUCCESS in 8m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7db47d8412824af8abef01b9192c261d : SUCCESS in 10m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a5c0f8c8dde94429805b2d972a445a5d : SUCCESS in 13m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/062aadfab57146a1ace3b0ad3d924aa8 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bde27fbccd6a4b469ea6e267467bfb68 : SUCCESS in 59m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fafa9c7c1774c2f8d18f143b3320d37 : SUCCESS in 1h 11m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea05d925bc58440e91977d716e536c08 : SUCCESS in 1h 29m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e43f24d32d014ebf98aa25cb180c28f7 : SUCCESS in 1h 09m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5f051218be739bb78a1f40d9892ef66eb7d877d8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-03 22:31:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d75eeeb895e66f1bc1e11aec86050138a8960bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 02:11:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad4909b6201465c8569c7304fa4c5fe : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6070118d1b400eaf2432439f7aef07 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4da3e5370c4e46f08279abf719c36e07 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e397361e2ac4a369e63da42d8f12102 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e51f1cc5396e4e369ffc69af442b4fa6 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0166d84d29041b085fc62ceefb0c6a5 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7eb9125bcde4260b0b4c520645ceaa7 : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/17a8f9fab9ec4aabb2e60e770066946a : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7c725d9eb2a2474e92ff7a7daca5727a : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/83757f3555c446f39ef39b50db570790 : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f1fbd3aea3e41b6a1c3252aab7469e2 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b9db1e682d3648b491174ada5b27336e : SUCCESS in 9m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a4900dc9d4044166bffcefbee5d656eb : SUCCESS in 12m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2431adf9f6494c8cb90a16c22c6a59b4 : SUCCESS in 27m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c7503683fec34ae493c3747431aca3a9 : SUCCESS in 12m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6f45698cc6e64cd0912137c39a90aced : SUCCESS in 17m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba2b58c890bc4454981c0ebd12eef7cd : SUCCESS in 20m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5381856c73654df2bac225f4a4633006 : SUCCESS in 6m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/835902cf895940dfb87fcddcf0ca0b8f : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/132ef89d94424774b6b8acc86dd6b780 : SUCCESS in 7m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c963472a5dab438484eeeb9fd516fdab : SUCCESS in 10m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d0aadd29b0164358bb54ddc8c201b1bf : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19f6aa1f80464edd99063af1a0e45545 : SUCCESS in 1h 05m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/16262bdc15f2482dbf394efdd0a1a898 : SUCCESS in 1h 25m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39cf581d333c49ce999421311f06383d : SUCCESS in 1h 32m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fe561d1de09045818a5c148a65ee7adb : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ce83090695298fb4a0802dd2c9defcab0d262d98","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 02:31:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e4b5f11edf8ac4b301c9276e2674448ad520980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 02:31:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0840e16913afbe87130592f17021cd7395c2bf69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 04:08:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3593a91e2bf142169433aaa8b1f18629 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c9124380fd4762a69ec7a4a12f8e5c : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/839a97b729994934906032a97dfbde0a : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41cdea0933b74df0abd71aabc876f315 : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12bc2f55f39c4395a0e685cb7f35f303 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32f9652f47214e57881d19b23466cf72 : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d1f4367436564ae3878f706ef6d3591a : SUCCESS in 7m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9d202af9c37c4c67b88ca4a3c1accd45 : SUCCESS in 8m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6d32808189b14f849b58c04c09eb2bd9 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ef9a7c401f94db1afb89b631dec1b8e : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7abd78971ad4a5d906b6ace98a489f9 : SUCCESS in 7m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a180be1a2d294839b7bcc7760a01a689 : SUCCESS in 18m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/da899bf768214ecb949a2e46b473fdee : SUCCESS in 25m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/685509cc544f4da194f2b3bb35a16e77 : SUCCESS in 12m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e4f5dd63ca4d45dba685e9698e44caef : SUCCESS in 16m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc758b564dee414bb6be515a2e672cf9 : SUCCESS in 15m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/51c41e1d577a4899830595eb566bc5e2 : SUCCESS in 4m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e7c59ac320624774aa3729d90452186e : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b885598375924b51a2953098f29368a1 : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ca5e2670183b44398f5e8597caea7fbe : SUCCESS in 10m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aca1ce5e3b82457ba89d07d0f4c960a9 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfc15e4cb9194e3da4ee104572400df3 : SUCCESS in 51m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/883c5aa92ffd4a38a13b93d56c4ff6fe : SUCCESS in 1h 04m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8014edabd5b4425ea2a68977156c49eb : SUCCESS in 1h 14m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bc066a6f4e4740d18d92db4b6761c787 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d55f24e3ca25ca634eaac5898dacc796ffdbbd25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 04:08:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cc1f8a3fbf2a4b69b11e8b87f116ba459eb95ec3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 04:10:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a17f2ae18e14be39775f65031dd23d7 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bc8626d51e9db66a267d7c6dd8516b1091c0fc3","revisions":{"9bc8626d51e9db66a267d7c6dd8516b1091c0fc3":{"kind":"REWORK","_number":1,"created":"2020-03-03 14:33:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/711027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/711027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/711027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/711027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/711027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/711027/1"}}},"commit":{"parents":[{"commit":"b61ff7eba2bcbeeb3fc743e14dfcdbfad4b4d99e","subject":"update constraint for python-glanceclient to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b61ff7eba2bcbeeb3fc743e14dfcdbfad4b4d99e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-03 14:33:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-03 14:33:56.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 2.1.0","message":"update constraint for oslo.cache to new release 2.1.0\n\nChange-Id: I0b0191e57d9dd4fe144128e1c683d307098fe144\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I8ff829a8884688d23d3c8298b3c09eb484148e28\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bc8626d51e9db66a267d7c6dd8516b1091c0fc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bc8626d51e9db66a267d7c6dd8516b1091c0fc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
