)]}'
{"id":"openstack%2Frequirements~711028","triplet_id":"openstack%2Frequirements~master~Ic36ef4c414d6442fa597c2a358247e117b2c9d73","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic36ef4c414d6442fa597c2a358247e117b2c9d73","subject":"update constraint for oslo.vmware to new release 3.2.0","status":"MERGED","created":"2020-03-03 14:34:38.000000000","updated":"2020-03-04 00:23:03.000000000","submitted":"2020-03-04 00:18:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711028-1583281133247-bef65be0","meta_rev_id":"9fd40df3b98e86e92f23ea6c2486c01666f771b8","_number":711028,"virtual_id_number":711028,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 00:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-03 22:31:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-03 22:31:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 22:31:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-04 00:18:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50dec290cd4e03994ea910da113f937feff5535f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-03 14:34:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03ad93d8904412780a587001fc3e80e104be3568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:04:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d2b8a6c9e4411ca76e272328aedefe : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7189985d0a9d479a8ee4ceb0ea7c0fb9 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef84583a9e9948fb8767fd0056e9fc8e : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eaa31e4ac60b470abd38c90b9db05844 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99f951d814744cb8889ffe1f2bc44c24 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/636fff35e64e4cffbe19f800ccfd11ab : SUCCESS in 6m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a4c26e7428d4acc8b585a710ba65dd1 : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/953c578ae89c446e8d37d2563d5f5f39 : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4dcf9f1ed2df4264b3004d9cc2a343d2 : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7c763a8c1e694b7b9b72693d3a659305 : SUCCESS in 7m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/90f978d8b6144b00b528deca277a3a07 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/041d799ee04d414d84298daf86659b56 : SUCCESS in 8m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1ddcd7a5347a4b4a87556b94c609aaef : SUCCESS in 24m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/37216cf2f5434cef9b82a2520fbf4bdd : SUCCESS in 26m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c5c72ade1d764d08865093d5eb87e8fc : SUCCESS in 17m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6bcac1635f0b459dbc358e436c520e0b : SUCCESS in 14m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e2e8921e73e4a7ca45ce9567ac69e09 : SUCCESS in 19m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/37597b49185c4471be81ae0c8193e5b1 : SUCCESS in 6m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/63e0eca59cd949baaf2faf49542f7a71 : SUCCESS in 9m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18007ea54f7a4d0982e0d9e9e33587d5 : SUCCESS in 8m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e0d99d3029b64116842535df65cbf123 : SUCCESS in 10m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/21b3cb8e85244803a2d12b644b52eafb : SUCCESS in 11m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/365235b6b94543158229938ea82b563b : SUCCESS in 54m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd6a9aa12a6d462aa4844d24bdbb88ed : SUCCESS in 1h 08m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34605855ad38447ca6348fcd12580a86 : SUCCESS in 1h 29m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/32aeac748774471aaf13b54f88fe835f : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":1},{"id":"40d6bd062ef60a56277f71b5744bde20a8b9408e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-03 22:31:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8836372e06e93881fa9fb232eeb4646fcf0eccaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:31:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aff611d771e6d799b30094b957dcb88c363db335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:18:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3bb65e4bfcd453a873fa835ffc56d77 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf32681bfa94f01af126fda0242cd69 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d95ebc2a0a964854b0bceee81a9017bf : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/744e50fd3bbf4e67b1fcf4804b6e2cc4 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b5c168b6f0642aebb2ec52732959724 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5af7da85cda1484abba445bc8be76c7d : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e745a822114477ab3fd91a380fd2923 : SUCCESS in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/980b94373f6d4ead909e016cb547ea0a : SUCCESS in 11m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b003d201256741929c935367ede3d248 : SUCCESS in 7m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f20ad085dc0d4c2b882e29d23a070ae9 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/68057f244e004be592201488dc7f2425 : SUCCESS in 9m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0b268f953a5f47eea2d2b996e020aff1 : SUCCESS in 20m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02a43ef06fd84b8fb39483ae45bfec12 : SUCCESS in 22m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/520d8f7e192d47ffacaf61f3f950fe91 : SUCCESS in 15m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a5a41295256748d4b3d257a5c9f29f5e : SUCCESS in 14m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74860a8143b84fab937a08a1f9c1c6ae : SUCCESS in 20m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1aaec8d6cb3048748c91f155a946c27d : SUCCESS in 6m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9268acbbf8e74573b2cbfca1d5fe1748 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41d727829d3d41d887c316bf39c99ceb : SUCCESS in 8m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3eac718441074d22b914ba99c836b750 : SUCCESS in 10m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/429370db0eb74880b5b57fa29dbdc815 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b4c632eee1c47e6a6205dbb7776d6a2 : SUCCESS in 53m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a4316abef794cec806c966f96fc9756 : SUCCESS in 1h 09m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/378e6a0ebaba4f9c848ad14cbbb9b3b4 : SUCCESS in 1h 22m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0ae0b78f7f8743b49307af66281e9ca4 : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":1},{"id":"1c6d126d08b679bb189e5335e66eccf0ad84810e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:18:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"20c745ea3175edb1c41c8a43b7001cf756617ca5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:23:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7342cc9264040e4bd522fda767291e7 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a891afe6f7e648a79ff92fa0065693f45ceb096","revisions":{"9a891afe6f7e648a79ff92fa0065693f45ceb096":{"kind":"REWORK","_number":1,"created":"2020-03-03 14:34:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/711028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/711028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/711028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/711028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/711028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/711028/1"}}},"commit":{"parents":[{"commit":"b61ff7eba2bcbeeb3fc743e14dfcdbfad4b4d99e","subject":"update constraint for python-glanceclient to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b61ff7eba2bcbeeb3fc743e14dfcdbfad4b4d99e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-03 14:34:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-03 14:34:36.000000000","tz":0},"subject":"update constraint for oslo.vmware to new release 3.2.0","message":"update constraint for oslo.vmware to new release 3.2.0\n\nChange-Id: Ic36ef4c414d6442fa597c2a358247e117b2c9d73\nmeta:version: 3.2.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I8ff829a8884688d23d3c8298b3c09eb484148e28\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a891afe6f7e648a79ff92fa0065693f45ceb096"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a891afe6f7e648a79ff92fa0065693f45ceb096"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
