)]}'
{"id":"openstack%2Frequirements~711029","triplet_id":"openstack%2Frequirements~master~I496f56d719077e91ca197c600e06e0859eb91df9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I496f56d719077e91ca197c600e06e0859eb91df9","subject":"update constraint for oslo.serialization to new release 3.1.0","status":"MERGED","created":"2020-03-03 14:34:42.000000000","updated":"2020-03-04 00:18:51.000000000","submitted":"2020-03-04 00:18:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711029-1583281131837-30632db2","meta_rev_id":"a8925195931e20e7e7e5f1f2928cef23830e0e97","_number":711029,"virtual_id_number":711029,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 00:18:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-03 22:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-03 22:31:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 22:31:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-04 00:18:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f97ebd5d67d4db4b4214086fc21e1c247b5cad2d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-03 14:34:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"759e60dd62b9a6b67a4017d3c9b339cdb0557843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:03:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc06af3d865742d193a9e4da84bd59b6 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3669a16edfc048b29abf172e05f1cb30 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88b50b0bd54342cc983d2d1b079295f2 : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff124a48b7c349edac9deb6e207a9713 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d28a3338ba9f473f8cd978bf65427324 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e8c5f681acb4318bb444ae1c34afd9d : SUCCESS in 8m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac28a8874ec3421cbe53bc7186cff8cb : SUCCESS in 8m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f35fd85a4dc4418d80305c0b904273fd : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/578420a2372c466983aa9a705ff02300 : SUCCESS in 11m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a20500e23d9549a8af4652debb540666 : SUCCESS in 6m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6eaaebd00c0a4b7fbde2541eaa8ca609 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc04ad6f1c9a4834896a936fdb879a2f : SUCCESS in 9m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a9034dbc66db4517bbcb321d93d2d794 : SUCCESS in 14m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cd90e23de59043b78c732dbdacc36b9a : SUCCESS in 32m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7299e4dc024141cdb9a0ff2d5150bdad : SUCCESS in 14m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f90fe3028e404c758243043f4833eb21 : SUCCESS in 17m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce23233f0a8a4e7b8d806bb7e74cfeda : SUCCESS in 18m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bc85d15f4f7f4e828192c12d7738133c : SUCCESS in 6m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f7f8a4928778489bb5865cc82847fffe : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/effe357d4bbf44d69b4c0e91ee6e2545 : SUCCESS in 9m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f2dd3c205e6e41daadcb86ccb0f056ac : SUCCESS in 10m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2d0f7514772e4b958b5eb157f42248b6 : SUCCESS in 11m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df21fd6c98a04919a0d54a236d6bc7a3 : SUCCESS in 1h 01m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/64d8d8b064ed47ea8272953d225e08c1 : SUCCESS in 1h 15m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a15c36793fa458d9c6a216382146d09 : SUCCESS in 1h 24m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0416905fd73d479d91b754816703f4b1 : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d6e4aea82f8dd5c3ad4907dffbf2183837844e63","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-03 22:31:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30baabfe2981ae6c10e299d2e639fcd1dc30fa92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:31:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f79e9659742cf1b6fc7ff9991940047616d2c78f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:18:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d802d787ed4fc12ee82aa5fd9a7eae51d5545bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:18:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71e0e72b03445f89fa3602005869061 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd12c34f38a42d38bb45b97f5e2f3b6 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bf664c6c3304999acba345944f30c53 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/108073705c7c409ba63812dd48bb4480 : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/965b47c4fb69435684fef96af4cbe1ea : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd46a57b24934537bf2af963d335fb5b : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/40bb547191f2472bb681088601e0b634 : SUCCESS in 8m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c12858a654e648189cc3153e33093d3a : SUCCESS in 9m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2e2866105b664e4f816dbe3b44feecd7 : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/58ff8fed93d241b2b282ea7174c9f16d : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4adad631e0ff4048958184ecdcf5c3b4 : SUCCESS in 7m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b21b251e68134c5c91fc0aa64ae81e01 : SUCCESS in 13m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/da1f614f74e04f20bfafe45bc0c4c48a : SUCCESS in 23m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5334398565f74920bde8c95f86129f24 : SUCCESS in 13m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/accf87ce32164ed5ae74b57d4729cb6e : SUCCESS in 16m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1c2431eaa054a459ef2471b87d91fe5 : SUCCESS in 18m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b6e53d0fa32a43d4a71f6ff8c71f8408 : SUCCESS in 4m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1d69a4cbb2e644188f353570bf2b097a : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e6b86e97f38f43858dc7533e81e499a7 : SUCCESS in 7m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/36dff1254e0741e1be1704f4a2c757e0 : SUCCESS in 10m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e0fdbfdf0cb94c24bb9508beba435216 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/253de43060b6461facbe89646e316774 : SUCCESS in 57m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfe15a3fd5cd4304b15f807ee72d95de : SUCCESS in 1h 14m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49f93eaaf0084088ba84dc1d24d22297 : SUCCESS in 1h 23m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ce0551ba9f9742b59d43a79ebaee08cc : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d4c58f4e95899a1c81209680c24cefbeb1a68b2","revisions":{"5d4c58f4e95899a1c81209680c24cefbeb1a68b2":{"kind":"REWORK","_number":1,"created":"2020-03-03 14:34:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/711029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/711029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/711029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/711029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/711029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/711029/1"}}},"commit":{"parents":[{"commit":"b61ff7eba2bcbeeb3fc743e14dfcdbfad4b4d99e","subject":"update constraint for python-glanceclient to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b61ff7eba2bcbeeb3fc743e14dfcdbfad4b4d99e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-03 14:34:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-03 14:34:39.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 3.1.0","message":"update constraint for oslo.serialization to new release 3.1.0\n\nChange-Id: I496f56d719077e91ca197c600e06e0859eb91df9\nmeta:version: 3.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I8ff829a8884688d23d3c8298b3c09eb484148e28\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d4c58f4e95899a1c81209680c24cefbeb1a68b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d4c58f4e95899a1c81209680c24cefbeb1a68b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
