)]}'
{"id":"openstack%2Frequirements~711030","triplet_id":"openstack%2Frequirements~master~Iac36ae8ec7c3defc941a94a855ae68b2f05e9499","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iac36ae8ec7c3defc941a94a855ae68b2f05e9499","subject":"update constraint for tooz to new release 2.1.0","status":"MERGED","created":"2020-03-03 14:34:53.000000000","updated":"2020-03-04 00:20:55.000000000","submitted":"2020-03-04 00:18:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711030-1583281130377-72e87165","meta_rev_id":"df463e7ca9b003c1d95349147bb36205836d2472","_number":711030,"virtual_id_number":711030,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 00:18:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-03 22:31:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-03 22:31:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 22:31:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-04 00:18:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c92329ddf4af08fe491bb6c8785284604657359e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-03 14:34:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98ea0d477acc7dcdce86700a23e3c1d9a5aeae06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 21:56:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/751504ba7c7544edab7bfe284c4301a0 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ec4e991b66421c93ffcd55d7c7dda2 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61779b5f0f1e4e6886389532fdb487ca : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/515376df8f0842de9ad134b3ae3f25e6 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcac3137d14f45a6b04ddb116752b335 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6bdfe74dfb44c9db5c06e362d17bdfe : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ff69fbc0dd149eea765dbc7da855bdf : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/699583da3801417baaefdf4c679eac6e : SUCCESS in 8m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cfb0d82951064a7e9618e25dee2989f8 : SUCCESS in 11m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/21f78815c2c04aeab339207b1485f0c9 : SUCCESS in 7m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8073f4a47b974eb897575d888d185307 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf95758886484e5780121d8d1902f356 : SUCCESS in 8m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/60c8cdafbddd42edade41c14233b28a2 : SUCCESS in 17m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/85745f5e71f84433ba6fcb50e239ef9c : SUCCESS in 24m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bbdb2f00d71e4447b695ed473654294f : SUCCESS in 13m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/99ae7f426208435cb87c1eb84bb1e0e5 : SUCCESS in 14m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8b03d634b984342bd59b5ec76ec15ec : SUCCESS in 17m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/243540e01ffa4055acadd0f59d3ac25a : SUCCESS in 6m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fdcfab47dc9a43ac9d309cdb2c588f7b : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05db66d5e2584ae49e03b695033b232d : SUCCESS in 8m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/11981ce2ed0a4f08ba9f17743b51d184 : SUCCESS in 14m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8df2fecbfaf6449b933790ea472eea1f : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa2f079da1524cffa245dd9c824a8908 : SUCCESS in 57m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e08c7b272e9441af90fe2e0ad06f528c : SUCCESS in 1h 08m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f93da9f3ec9b43bead53b7b23b9e3667 : SUCCESS in 1h 18m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d334eaea1ae942fb8e2e8628c45bd9d0 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":1},{"id":"129ff29c214050717a5dc3b9818ab50ec2df1ce7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-03 22:31:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34c9d1f43293e312e7ec085dcd246db10ee2e353","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:31:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0cc08d288771de7d07a921f7da5179a7fdbacf5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:18:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce8401430b34065b7ac20a852e8cbe3 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc9046ee6ead4a77b9b538500040c25f : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c16fcd14c0b44a688741d1bdd3ea44e : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/412e69c4a8e1405faf10b9797ec3f00c : SUCCESS in 5m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce750f23b3094076a633f012780a8e72 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e79aaf6784e4b8099b81c14d0082f7d : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0db0de0f70884a44b8475259e939b5f6 : SUCCESS in 7m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f9e79e69ecd04a768a12841bbfd14881 : SUCCESS in 10m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/79196520bd1a4a02a02a703e2b7771f2 : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2e0b520f1ddb4311ae70442835261949 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/77a5b61965434426a5e08bed1fb14e4c : SUCCESS in 7m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a1df638ace834d149175baaed207a3a5 : SUCCESS in 19m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ea62768ce71b4d6095829612cd04090a : SUCCESS in 27m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0cb3ff379c55445d8a5aa256aad79828 : SUCCESS in 19m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e72f04f7551a4408a1ef3d381f8c1d44 : SUCCESS in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc72dd43e8e745de9c129a4937c28909 : SUCCESS in 17m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6d0d6cd9fa334b6b8a647ffd7b19f681 : SUCCESS in 6m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f0fa2e972b7e481ea32f6badb9bede55 : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e647a8c9081b497e98077012812c1fd8 : SUCCESS in 9m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bcdc3afc531f47a8bce4bd55efc76b8d : SUCCESS in 11m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6de3fdd7abc74ab491035c9ec82c3b8d : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/621ad274794e4b99a48ff57ad639ef94 : SUCCESS in 53m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6fbcea2e0c3348cbb659372380197df3 : SUCCESS in 1h 10m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/484d9f33b1a741c297cfb4105ac2b871 : SUCCESS in 1h 42m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cf2fa4caa151466b87267a2fad58009e : SUCCESS in 1h 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"15f5a978c48895c0f816e526bbc71f59d67187eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:18:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"216be804f8735dfa7ccb9721b2dfeacf92846e46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:20:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab744a550a9f4860a84385c285d4967b : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fbfc813119c9e21286c538b217955b4e85f868f4","revisions":{"fbfc813119c9e21286c538b217955b4e85f868f4":{"kind":"REWORK","_number":1,"created":"2020-03-03 14:34:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/711030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/711030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/711030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/711030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/711030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/711030/1"}}},"commit":{"parents":[{"commit":"b61ff7eba2bcbeeb3fc743e14dfcdbfad4b4d99e","subject":"update constraint for python-glanceclient to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b61ff7eba2bcbeeb3fc743e14dfcdbfad4b4d99e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-03 14:34:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-03 14:34:51.000000000","tz":0},"subject":"update constraint for tooz to new release 2.1.0","message":"update constraint for tooz to new release 2.1.0\n\nChange-Id: Iac36ae8ec7c3defc941a94a855ae68b2f05e9499\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I8ff829a8884688d23d3c8298b3c09eb484148e28\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbfc813119c9e21286c538b217955b4e85f868f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbfc813119c9e21286c538b217955b4e85f868f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
