)]}'
{"id":"openstack%2Frequirements~711140","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I2d19de52263c844afe1a7a2e02be42084cc86b95","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I2d19de52263c844afe1a7a2e02be42084cc86b95","subject":"update constraint for oslo.utils to new release 3.41.5","status":"MERGED","created":"2020-03-04 01:12:19.000000000","updated":"2020-03-04 09:58:24.000000000","submitted":"2020-03-04 09:58:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711140-1583315904054-99049c42","meta_rev_id":"5682c6173e1b1fea08aa3f3b1bb72c2ab52efe32","_number":711140,"virtual_id_number":711140,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 09:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-04 04:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-04 04:44:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 04:44:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-04 09:58:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"424eb65c4fe9dee6be64395a8510681465e671cc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-04 01:12:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d1e2e1900360c97521f717cb0682efcf5004f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 02:46:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c49a8ae80a7641ef8b9d75ca21cfbaa7 : FAILURE in 3m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fe67c9b546040ccab30480064d1a8d7 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21ea1c0562cc4524b6aa02d1c394cb47 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c19fab9718f74cdc9d6196cacef57ed7 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a9ca0f12a824f78af14e0e533ea1665 : SUCCESS in 6m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a90d04414f64868878a1df6271474c7 : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b3a2f8ec34294baf8376c9f8106d8f90 : SUCCESS in 5m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e2ddd34384c44eb8b255c1e8e5224572 : SUCCESS in 8m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71a794a334674d49aa3b0be8e4bd29a5 : SUCCESS in 8m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/cc2bdae61b8d4ca4a2b180beae4adddb : SUCCESS in 12m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c98b08cf0a8547798e8f67d2b5f72d94 : SUCCESS in 10m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/96ea0e52865449c48094d095381c1020 : SUCCESS in 5m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4f47f719b1d14d9ba0a314c44872348e : SUCCESS in 5m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/91be92ecc43f4a8cbbf7eb3ea55e5932 : SUCCESS in 11m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54ea888367b64e1abaa98445dd765f79 : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a218429b6d145be83f0103d9fe395af : SUCCESS in 7m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/21c9d1952a154cb8ab5450de7e66bdff : SUCCESS in 24m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1df78c06758b4405a6fa63e61038703b : SUCCESS in 26m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5a4a4dbb99854980914e400975f97725 : SUCCESS in 23m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8f6bf37262e24d09902ce932665ba9c3 : SUCCESS in 20m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d69d93ab866848e89716b77b6200d9c2 : SUCCESS in 14m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9f8b1a3cdda242d2a6f8807bee809bf0 : SUCCESS in 15m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/dcbb41ca4882427f9d49ab628d9f6090 : SUCCESS in 13m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b86314dad7d14967a10bd61858ceec53 : SUCCESS in 19m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da805e85e00949458669da577d24372e : SUCCESS in 5m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cd4176060ab24dfab6773f3405be20f0 : SUCCESS in 11m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/afa406c38cbe4b12a5b1652892e12b18 : SUCCESS in 9m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/05b10a40d3ec486db4f43921ad6e4b39 : FAILURE in 43m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/00ca682a8b004183805160489784f35b : SUCCESS in 1h 27m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8e5970a669b1445bb505f9a203a0d4a2 : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6f0c93e1b97840301de6baf07fb5cdb4d89477a9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 02:57:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03fbf2ae44451d9aa254baf7926a92d92a5173fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 04:41:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9adf7cbe0a5d4572b1ba6bc1a42f2d0d : FAILURE in 3m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0948e8ba5d549928e95ce8d125f4b79 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c1368a1548d4b17af3d39fce8aebf84 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b5ab1eb6acf48348a15d5e35034f961 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2adf7cf67ae4f00bdc85c8bfc2b87c4 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/519a97b16ea0425191df6621d14d06ac : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/300f87be476e4bc2bb25b63ce49a04da : SUCCESS in 4m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7d73d4bb377344d1802004d04a377335 : SUCCESS in 8m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a7f92f40cbc84de0a49e745d918c6c1b : SUCCESS in 9m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e9f2c18de87144b6aad7ea1a6981b0e8 : SUCCESS in 10m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/be21c43d4ca24e959a812a830b71d52b : SUCCESS in 9m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/47a500eefaa34d25a46722eb3ddd8f32 : SUCCESS in 4m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3b22d04959c448d9a05775a7a463d643 : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9057f566124d4ad09ef178066c55feb5 : SUCCESS in 11m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8f07e2d418cd44378bd219d0f23e3de9 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3c8e39395474642a45d4f5346a1e2e9 : SUCCESS in 8m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8854075a89bf4b75b113ee0d6e320218 : SUCCESS in 13m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/817c7fb8822147629376cc00e5029ec1 : SUCCESS in 15m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6b3143894f4b413994c17f4609c504ce : SUCCESS in 25m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/32433a9e5f054bc3819ea0d06c5f8ce8 : SUCCESS in 23m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c862fa37cc904dd5a251bcfe8c057f5c : SUCCESS in 13m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/650b8f1528fb46848389a0a3903e6761 : SUCCESS in 14m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3db493ebba5d44c899d1168d1c3d92e3 : SUCCESS in 16m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e72dbd60e7e24ee7bc03a53740a4ca7e : SUCCESS in 21m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95e36dea00cf4fcb9c515e0e66b58ea8 : SUCCESS in 4m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/899e173747d349df902c0ab9c66e10b5 : SUCCESS in 9m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d4aa465d6cf54aa5ac3b6d5b73561fa3 : SUCCESS in 9m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b94da86c5ab3449e90d412d775cd51c0 : SUCCESS in 1h 11m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c1e875c802a24f2b90960e97e181314c : SUCCESS in 1h 38m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3250cf10fd7b480eaf00bb29f80c325f : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":1},{"id":"15e3c2348d7217874fdd88152da80b7d9affb134","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 04:44:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e3a8f5349264c305cca929c590ae0e5292b2a70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 04:44:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac9b0e2e64399daeb0078ce413057009d4ba85e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 06:26:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ee1f88e6594f85b2da54830b086193 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86b44a88450e43f9ab91d521cbe73f66 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3980b5073e74249a1c4229fb08c49ee : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/008e9a4382224cc48d2500a9b556b5f2 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3eae270e9c548dbadc0703c0f84b922 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e88b14698d44da3b4f7a9e921488927 : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/27b5bc9dbf66430d8c77b57b088b951c : SUCCESS in 8m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a0e7a17c57049efa4fd688b42d14f59 : SUCCESS in 7m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fed560dc87734179bd09fd538f73378a : SUCCESS in 11m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/48e4b11d2afe4841a140a38542ec0138 : SUCCESS in 8m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/60c173806b344a14b93e13ef3e77b3ac : SUCCESS in 7m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/048d7784441343288ae3e853b80355d6 : SUCCESS in 5m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/00926cbce8d244b294e46434d3a75715 : FAILURE in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/84a6c8926a1040febddf9f976b405da0 : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b71434d2df044664aa6c56cb1fac8338 : SUCCESS in 10m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/158a1500735d4648b793cc7d3e8d9a5b : SUCCESS in 17m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dd66a3deef9e4a5aa0fd01b3e0173102 : SUCCESS in 17m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/62dc8f65a42b40898724aee33c034b7d : SUCCESS in 19m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f71f332b835f44d0bfee48323444a684 : SUCCESS in 29m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b17cb32ba4d549c4b22549690727b84a : SUCCESS in 13m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/770ec06c199546749c207de994656f7a : SUCCESS in 19m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7adc093dc9af4a838315e8eea66ce120 : SUCCESS in 13m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a435a974d436468c8ae2e1da39a29a92 : SUCCESS in 19m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dae79fe6e7cc4110840166d52001bee1 : SUCCESS in 5m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a32483ae4e68433c8d387202d26052b2 : SUCCESS in 10m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c48611317e984775a15549af6724f951 : SUCCESS in 11m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/915f1b198b034e89b85c43750d0c940f : SUCCESS in 1h 02m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0cd6a2679ad046c8b6f15f4a3d2cb27c : SUCCESS in 1h 29m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5a39c22d755a4fc6b1e26c0f03c60ccc : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":1},{"id":"32275567c5891438e0725de87a30fcd655b519a2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 07:02:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d16c3c3d1c7be486e172b647d9a60bebd23fc41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 08:28:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/78f79a4597bb49b8ab3f60f48e42018e : FAILURE in 5m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f0d8f85b0c4b5686f9e1d53003224d : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c5573fca91bf41e48a6a59759b7eda02 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85f21670db874edc914ab6e1d8dec654 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/270c78bfa8844ab096f5f875aef6b982 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f6b5d19a83d4d2a8417f0e88fa33133 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed5424fd25c74513b89f583fd13fb7cb : SUCCESS in 5m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6540faabb121451683d3de8e4141a1c0 : SUCCESS in 8m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/143b9b925ce244bcbe223e39bf247b6b : SUCCESS in 8m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/62c8b359768a4e6c9846e1a22eff9ad5 : SUCCESS in 12m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ed22cc7bfef34cc4adde00ee8bfd48d4 : SUCCESS in 10m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5c947f81ebbc4251a9600f3c3248bd57 : SUCCESS in 5m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b33c3df7e48b40bbb7232a8ffcdd6c06 : SUCCESS in 8m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/11909bae14d4469a900af46413c839db : SUCCESS in 10m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d3595e618544f90baa1e36acd682e5f : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2365d56a92204b86b2301097c141f861 : SUCCESS in 8m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b89e97648d08472b90641036f7436e7d : SUCCESS in 15m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d4487bf92d604182a6d730890095a6cd : SUCCESS in 21m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3e8b901690fc48b78eed6e21019773bd : SUCCESS in 28m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c97d349a829940ef81bd59646783baa0 : SUCCESS in 23m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/01637e568d964cb4b329fb69f128af39 : SUCCESS in 17m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1a5f9e49502645a29fe8943362541855 : SUCCESS in 17m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8547ad2b0c1e422282ca33908f25054d : SUCCESS in 14m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dbcdb5828af84e7fa9c39f7bf66a67bc : SUCCESS in 20m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2af5d500053445e79837b91975c3c4c1 : SUCCESS in 7m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/78df20dfa8044a6b9bad6f08805e85e9 : SUCCESS in 10m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/714f7f4158d54e56b506b7f094db3347 : SUCCESS in 10m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be397a38471a4f64953f25aef327958a : SUCCESS in 1h 15m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fc34c2076fe94a2196b22e3657f99fde : SUCCESS in 1h 24m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7974eda220a54aab9d8525c9625f49e4 : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":1},{"id":"53c2ce99bb3452830bf190d05374be5302633db0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 08:29:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d0560b8a9f20e122b6b086dfc5c27820db78516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:58:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9608dd41f845cdac559283d2515e11 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2eb5e48aa2d04b8dbd92ba73bf8bafd4 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0102871d85b4874849c67c451f57e24 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81f975318aa74c77907fc246a5d8c705 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a397d76e4604fa7b37302f8d464f2eb : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7fb3be20706a4bb391ccb15d37223ed2 : SUCCESS in 5m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/675135fe40804075a5e2125fb8dd35d4 : SUCCESS in 9m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8dc73467b3924f6c8bc277440ed08015 : SUCCESS in 10m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6b63e920a3d74f25a633c2ae9d31b5b5 : SUCCESS in 11m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/98cb28e7310e40f68cd975bf66419ea8 : SUCCESS in 9m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0a544a0ac1f2403a8ea67685fbdd68ab : SUCCESS in 5m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ca01c1aae3304c57b6596dbb983a545c : SUCCESS in 5m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b8ecb6f4c90c431ab4fd12decdc85118 : SUCCESS in 10m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd4c7f561a8040918cc1f29b59a587cc : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4c785d4c0024c4f886e09365b682e82 : SUCCESS in 8m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ec1260387e6043919cf3c03fc97bc23e : SUCCESS in 19m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d93fd06f23f2491d8bc8ed972048382a : SUCCESS in 14m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/59a6dfbf0ae94c0f92a04b69d4b73f54 : SUCCESS in 23m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e28cd903f38f4d87b89a1b449483384e : SUCCESS in 22m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c96533d3868e440e93addbd12f9dba4d : SUCCESS in 14m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5998fd65c18049cd8b04476d30b6a8ba : SUCCESS in 12m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3cd659400a3f4713bab564469ff12173 : SUCCESS in 15m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e3adffa61f243b1b3176f1bb139853b : SUCCESS in 23m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b74da49407e74e6e952211cc7a173e32 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/09e219008e6e4d16bc198df52eaec4ab : SUCCESS in 12m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d93bea8130394c3ebcf7c09d3ee8e8b8 : SUCCESS in 9m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/046cfa0422bd4690acd5cb3e55771017 : SUCCESS in 1h 09m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/46acd88ce43549a3b1f61cf45bf2824f : SUCCESS in 1h 28m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a89a8f7968ce479680884eb985ea6443 : SUCCESS in 1h 19m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2964dc07ffa249c65982ad917dade306a137bb0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:58:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b1b94be547ac00d53fb80671bd15a1245db3807","revisions":{"6b1b94be547ac00d53fb80671bd15a1245db3807":{"kind":"REWORK","_number":1,"created":"2020-03-04 01:12:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/711140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/711140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/711140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/711140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/711140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/711140/1"}}},"commit":{"parents":[{"commit":"3fa235bb42386f704dfdd8e24e264f42fd8be386","subject":"update constraint for keystoneauth1 to new release 3.17.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fa235bb42386f704dfdd8e24e264f42fd8be386"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-04 01:12:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-04 01:12:17.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.41.5","message":"update constraint for oslo.utils to new release 3.41.5\n\nChange-Id: I2d19de52263c844afe1a7a2e02be42084cc86b95\nmeta:version: 3.41.5\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I46ab97401d60b3f03da7d4c3414f0e51cd11bb4f\nmeta:release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b1b94be547ac00d53fb80671bd15a1245db3807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b1b94be547ac00d53fb80671bd15a1245db3807"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
