)]}'
{"id":"openstack%2Frequirements~711147","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ifa43234369c9c5293d196b14a2ac4ec62f65d2c5","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Ifa43234369c9c5293d196b14a2ac4ec62f65d2c5","subject":"update constraint for python-cinderclient to new release 4.0.3","status":"MERGED","created":"2020-03-04 02:25:55.000000000","updated":"2020-03-04 06:15:29.000000000","submitted":"2020-03-04 06:15:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711147-1583302529533-99049c42","meta_rev_id":"aa99301358902e44ce8e141a8843d70818d57234","_number":711147,"virtual_id_number":711147,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 06:15:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-04 04:35:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-04 04:35:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 04:35:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-04 06:15:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19a2d3f8e55f567f520c9eb66453bb97854f0f56","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-04 02:25:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ddee5ec7ed07d24037fea2ff02f4a296867bf24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 04:01:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b06790256b16478b997413eaedd04bd1 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/895e3165351d4a2cbe42c54ef8b8f7e0 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84532ef9ca8b4852a1a9bb38317a27dc : SUCCESS in 5m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/06355564f5504e98adbbb360ed402434 : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15920cf202c04b97afee34fe45e36814 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e2c3e3e6dab49e5b9d0cbbe8877e69b : SUCCESS in 4m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2218aeed17844e9ab8903e26fda30213 : SUCCESS in 12m 12s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4a6bc7936cef428a94fe55f0ade87bbd : SUCCESS in 7m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5b7eba4756ba4e829e7c2f386a724f40 : FAILURE in 8m 36s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6b72bf2fb26c43d2b4d87d6d6d657cfd : SUCCESS in 11m 31s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/901fb3e985e844ae8bdc729162aee18b : SUCCESS in 10m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4cffa57099bf4a07b7d96bbcf088b454 : SUCCESS in 7m 17s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/79e76de3a121437b85057b3e194a53ee : SUCCESS in 7m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/18bca674c8bd4b0c8dd1b9bfd98c76ff : SUCCESS in 10m 24s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ec88dcc925ed4ae0bf7cdf729c467ca0 : SUCCESS in 10m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fcaa8bf9248e4e6587627d57f5eabc49 : SUCCESS in 12m 43s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b4a9b78f11f24025a6d49ce49a667b69 : SUCCESS in 13m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0a032b01587249289ddd7a9159fc71c0 : SUCCESS in 25m 18s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/8ef79150c65e4313a345c7586fb7afdc : SUCCESS in 25m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0e0bd7b528884de4a4bf432df26af7de : SUCCESS in 13m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3a33953fdd9f46358d052d1d718b4320 : SUCCESS in 12m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa88301d617f4aaaa8ad6f8394e592c3 : SUCCESS in 17m 15s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/6b65d092f2fe4ca08429efd4aa314cdf : SUCCESS in 4m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a86bed033a08415d9c443de4068fce61 : SUCCESS in 8m 13s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b359c744621a4b61a56848de831bc928 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/318b521e3db74f40a0287b4c0d48d77b : SUCCESS in 4m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c9b5cb931be43c69a19e49c42c7f482 : SUCCESS in 1h 04m 08s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/143667089a034cbca3cc2e868bc32d91 : SUCCESS in 1h 29m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cee1e705766141cd8a188414fabfaca7 : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9b62e0e4b6e6f5f739beb2c0c2b56bc1c3524474","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 04:35:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19e75ea49920a0adffa1d2ba077ed862be5de628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 04:35:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e15909ec76247d7717493e4994d14bc502fca6a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 06:15:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"958621af49ea073fdfe37858dabab26850dc4d23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 06:15:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3a86319a95d4efe94baef5aef578c39 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c33c18abd3cd4102a567d3120dc5dbe5 : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c03eb98a6d474acb80405da4293f5d2e : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/394d706913a744899d9b2fe849ca66c5 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76e1c4dd8f274bffa017b18d876a6333 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/88ead3773f5c4538ad27f773a1a4f5a8 : SUCCESS in 9m 25s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/cb2e090c5ec14eaca3cbb948e78ca05c : SUCCESS in 10m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/095bfcf6d17b4b4bb0711b0a778bb800 : SUCCESS in 10m 48s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/f84de7d9d4e944f493f1168d4f35fbc2 : SUCCESS in 10m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/93f4a62c502e46a7aceefb88db3add81 : SUCCESS in 4m 58s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/6883bb58a257435090e4bf56b4b24f18 : SUCCESS in 5m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/03c625f4fdc34dc08b9571b85117fce8 : SUCCESS in 9m 06s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/bb47ef65ea234decaa3fdcead145280c : SUCCESS in 10m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/86d413a67c7e4e08861173d41ce8369a : SUCCESS in 11m 59s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/104faf304a6c47c48fcfa98164427c67 : SUCCESS in 20m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9421ef4735f547eeaa166ddf566a323a : SUCCESS in 25m 06s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ed31c97978034934b429792c750fbea7 : SUCCESS in 24m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d833b28890f74b848467f190ad15efab : SUCCESS in 13m 28s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/0bddd9bd8a11481dac248b8199f51a7d : SUCCESS in 12m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ab20fc9083948c8a0cf809580e03aaf : SUCCESS in 18m 28s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/4cfe20d5e6e94e18884d77aaa95df1bc : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d09753850efe4a5687c6564387b7e84a : SUCCESS in 8m 20s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/d8180233e8e3459ea0fb2f8647206e7b : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bcf7b8d02de4df2995ab3bdf4947604 : SUCCESS in 3m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef817c64311f474ca5be632b9c5c1390 : SUCCESS in 1h 03m 21s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/a63517af8f06410694ba6cf9f18d8bae : SUCCESS in 1h 33m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1cb9d426c21a47f8b3993bf548e7457c : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0a534d5d549c6a4b5877377e91d714bdb7d55cb","revisions":{"a0a534d5d549c6a4b5877377e91d714bdb7d55cb":{"kind":"REWORK","_number":1,"created":"2020-03-04 02:25:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/711147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/711147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/711147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/711147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/711147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/711147/1"}}},"commit":{"parents":[{"commit":"568bc433d3bee63ea052b3ca352b4cf33164f021","subject":"update constraint for python-watcherclient to new release 2.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/568bc433d3bee63ea052b3ca352b4cf33164f021"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-04 02:25:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-04 02:25:53.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 4.0.3","message":"update constraint for python-cinderclient to new release 4.0.3\n\nChange-Id: Ifa43234369c9c5293d196b14a2ac4ec62f65d2c5\nmeta:version: 4.0.3\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Id4b1d189e578eac4d4ba5f02d3b45baab66da617\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0a534d5d549c6a4b5877377e91d714bdb7d55cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0a534d5d549c6a4b5877377e91d714bdb7d55cb"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
