)]}'
{"id":"openstack%2Frequirements~711148","triplet_id":"openstack%2Frequirements~stable%2Frocky~I8f3a47bd396ac2ffd7da5f30cf59269f671f06c1","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"I8f3a47bd396ac2ffd7da5f30cf59269f671f06c1","subject":"update constraint for os-brick to new release 2.5.10","status":"MERGED","created":"2020-03-04 02:30:50.000000000","updated":"2020-03-04 09:18:40.000000000","submitted":"2020-03-04 09:18:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711148-1583313520969-c83ad78c","meta_rev_id":"bbc6a82eaebd2ec9cf86e9cf2f4665e1f67cc919","_number":711148,"virtual_id_number":711148,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 09:18:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-04 07:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-04 07:02:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 07:02:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-04 09:18:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d91b224506994d088c7ba3a2faaa8d71250e54f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-04 02:30:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1b1c2740b96329398b816e0c4c54ee83d1da0d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 03:56:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0b70c2ea1c4c4cd680fd824575659dcc : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c16f8a8427d447890c0c803b2404380 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd3e463f16ab4355bb550783d37a467a : SUCCESS in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93e460697c1c4c6b95c1803ec68e4869 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00a500dcc114475a9dbde192d726da80 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/205170aa618341aaa7c31815435f44bb : SUCCESS in 3m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e3381332ecb14f48b2d6e26301590152 : SUCCESS in 7m 51s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d2ee8580eb18462fac5b976bd4bd7e1f : SUCCESS in 8m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8de8c83f1c664796ab2acf8e809ac019 : FAILURE in 8m 48s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/aefc2f24949841508a73df431f4cfa4c : SUCCESS in 9m 53s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/bde8bd09b0204a9dbd37b21f8ab8e70a : SUCCESS in 9m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3a4e218c8a7e499a9c814a4925b147d1 : SUCCESS in 4m 53s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d851f19f422142dbb73a82d408c33e32 : SUCCESS in 5m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e61e7f9d9ab349d19160cd2fef79b054 : SUCCESS in 9m 26s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/06f01644ae724a1f8654d070b45fadbd : SUCCESS in 10m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/293cc522855045d0aaa12058c356a6b7 : SUCCESS in 11m 46s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/eb6ea3d75b734e9f8e1f4f21bd937a7d : SUCCESS in 12m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c82bc55250014bc6b016cfe830c052ab : SUCCESS in 25m 50s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/53357d321a514d10be7d24600cf5d88e : SUCCESS in 26m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2838ee1020c94fd0b5139bd050c68af2 : SUCCESS in 13m 50s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/e3fae76ee01643688af14471ac9a7ab0 : SUCCESS in 12m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3380ca1811540d8925779f060d96953 : SUCCESS in 15m 59s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e775e76bd85249a5ac84db41b6ce4d2c : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/040711a750ab47ad85a431821abc9eb1 : FAILURE in 8m 25s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b0454d0378a64e13ac3007b5062ac000 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9b6960fadf447028e76908747fe07f2 : SUCCESS in 4m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1ad5b16cc9e4452b24c55223fb860b4 : SUCCESS in 1h 10m 20s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/55ccb2cb77574fc2951f07aaf80a55c6 : SUCCESS in 1h 24m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/65b888827acd4efbaf66820ad326fa4e : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":1},{"id":"667b8705d80099018f1f5b450bf5c64716535fb1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 04:20:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a69836d575e39f5c5d9f081147f352fccfa4be93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 05:53:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/872db9d476a34dc4b94e9fa1e6bd4d47 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cceab75f37545cf84ac1b577d208c8d : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/839dfc922a61431d803e4cff6afa3c90 : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fd69bf03a1094cf4889a89bad56695a4 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d289d57d523d4278b0090033ecab23e9 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff419c340747421ba8929bb035a3dab6 : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/76343dd7cc634e7c9e7333e651f31863 : SUCCESS in 9m 53s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/a50f51e0e92044a488b1cbe48ad1e7c6 : SUCCESS in 9m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/751218af85b345e88ee6216ef9a83029 : FAILURE in 8m 40s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a1615b8b11674e8d9cd76464b5049d7d : SUCCESS in 10m 15s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9f91aa82670c424ca7d0634c0def2546 : SUCCESS in 10m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9398683622014130bb318e5d0e507c66 : SUCCESS in 5m 02s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/6e1044be2c0448d0b96648218d2eef79 : SUCCESS in 4m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1fbc77df9da140f1803dd9ca9c280913 : SUCCESS in 10m 37s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/961354cec5b7426b9f2e235fb35eaccc : SUCCESS in 10m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/285c095f792d43dfafa7a40e963c265c : SUCCESS in 14m 42s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/c466a526446a44afbaa76b39ac570783 : SUCCESS in 12m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/155c83796f34468aa0ab840b7eae4e9a : SUCCESS in 23m 07s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f19e3d9f80a945b4a96d06d2aa0377bc : SUCCESS in 28m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e4e08396744e4c02b2358ac6b5f862da : SUCCESS in 13m 31s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/4638d09ea37f4f22a708fb27925f6996 : SUCCESS in 13m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8991a62f48be429f9976c15e912a04e8 : SUCCESS in 15m 35s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/369c539fbe2345e6a00d4b21d2a0b239 : SUCCESS in 5m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a7bf83f1fa574f549d3743b614aa0882 : SUCCESS in 7m 45s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/56e9e6a21f4a43c7add7183553e6b8e2 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5faa7ce1c36c4f1f8000c52eb46b6dba : SUCCESS in 4m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/965ebe10afbe4c979ef3f29a2fc270cc : SUCCESS in 1h 00m 43s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/39b2e132892147f285e02b3dd370caa5 : SUCCESS in 1h 30m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/33105b1866e44155b78f153fef2b959c : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1e5a0a5055ff1a028d819a5e72ae0b29e2155316","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 07:02:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea1d98cb72d45cfb8f564b79da77e2c4a323dcb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 07:02:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c5192cd141c031ea0c0f8d6fc6ea18ef9663119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:18:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d16a2a245b45ef87bd1aed6d7c319d : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2c2097905974dd48fc25ef52958a570 : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dae844ce5ebe49828ecfd65218cddb42 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18a17baedc0b472db880328583777f3a : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18f02151cd54419db75355dc29e3dc56 : SUCCESS in 3m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/358add4e4dbd4026aaecc52d9e72b4ab : SUCCESS in 7m 29s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/bc38e62f7fc246bfb5ad34674d041271 : SUCCESS in 8m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4846e98b0b744313980e47daa5492293 : SUCCESS in 10m 13s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/94d3a4534b844613bbd9907fc79cc41b : SUCCESS in 12m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6b64d4ffd8d14675a940d4916949348d : SUCCESS in 5m 33s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/af4a5df09f3642c2a0c863225d48c620 : SUCCESS in 6m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/619b0658523c40fcae0a090d73d4318d : SUCCESS in 9m 56s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/7caf47b7f2924f76837145b2a53ef247 : SUCCESS in 10m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4644e75db95e4166a42095c5368d09e4 : SUCCESS in 19m 03s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/cf07af4ec7d442ac8d9da4336f6531d9 : SUCCESS in 17m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/68633d93bda84bedb30149a981b13987 : SUCCESS in 28m 15s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/dec897558ad640b7b4615144825333bc : SUCCESS in 23m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c7e6a824158f4c67b5a5f1f4236cff2b : SUCCESS in 13m 53s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/df45d6ae6bc246a8bc89ad4c77d97993 : SUCCESS in 14m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a66b3b728d94dcca8d0353c9c361198 : SUCCESS in 19m 11s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/4e1b28a47fea46f88513646cb694d860 : SUCCESS in 5m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6d4b5b8f67de4d7bbb01f8041b7584b5 : SUCCESS in 9m 42s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/bc96fd720d334c23a8a97a4b9a55b5f1 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b416bfde1e8b446786be82610a625799 : SUCCESS in 4m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5873eb99ce44d9c88424b6ab96302a9 : SUCCESS in 1h 04m 57s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/b9b94350293d44f6b7e71ba04f8107d3 : SUCCESS in 1h 21m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1bc3216caa4242baa4b788302a3bd831 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ee74d605fdfc674de8e8933381cd6d6d7fd6c5f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:18:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e3c393d648937cde0de141619ea71fff2dd52b3","revisions":{"6e3c393d648937cde0de141619ea71fff2dd52b3":{"kind":"REWORK","_number":1,"created":"2020-03-04 02:30:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/711148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/711148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/711148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/711148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/711148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/711148/1"}}},"commit":{"parents":[{"commit":"568bc433d3bee63ea052b3ca352b4cf33164f021","subject":"update constraint for python-watcherclient to new release 2.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/568bc433d3bee63ea052b3ca352b4cf33164f021"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-04 02:30:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-04 02:30:46.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.5.10","message":"update constraint for os-brick to new release 2.5.10\n\nChange-Id: I8f3a47bd396ac2ffd7da5f30cf59269f671f06c1\nmeta:version: 2.5.10\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Id4b1d189e578eac4d4ba5f02d3b45baab66da617\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e3c393d648937cde0de141619ea71fff2dd52b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e3c393d648937cde0de141619ea71fff2dd52b3"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
