)]}'
{"id":"openstack%2Frequirements~711271","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Iaf759b183e6821bf6690195586e7e65bf0533c82","project":"openstack/requirements","branch":"stable/train","hashtags":[],"change_id":"Iaf759b183e6821bf6690195586e7e65bf0533c82","subject":"Requirements-check: use local repo like any other library","status":"MERGED","created":"2020-03-04 16:42:55.000000000","updated":"2020-06-29 21:46:42.000000000","submitted":"2020-03-05 06:53:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711271-1583391186905-bd11f162","meta_rev_id":"1e6a4340258ab30f4d16f80bd2c63ef784441b89","_number":711271,"virtual_id_number":711271,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-05 06:53:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-04 17:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-04 22:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-04 21:35:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-04 22:09:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-03-04 16:42:55.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 16:42:55.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-03-04 17:23:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-04 21:35:22.000000000","updated_by":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"reviewer":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"state":"REVIEWER"},{"updated":"2020-03-04 22:09:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-05 06:53:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"018c143c00f0f5b4af4b7ccfe941e7e66f132c9d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-04 16:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb3cee68f6937a0edd316520fe4c4f692c6bffb0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-04 17:23:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"466140f6280ff1856918c76fcc00e2145b62c829","author":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"date":"2020-03-04 21:35:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a42181845c51df18ecce5e6c8c3775beed6699b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 21:41:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/05e154e7b6b740eabcfd7745272c3db8 : FAILURE in 3m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7371d4708f54abbaadc7270cca83679 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2aba1137a814fcaa99291341b5b2563 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/285d9f22fb0a44248e0e41acaf748710 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/901cad80d948448db7b1d0ec7bcf0979 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f334397443948cd9a76958916dc5ff8 : SUCCESS in 4m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4eac324c85b1463a8629fc2a1c2e0779 : SUCCESS in 1h 16m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/239af58dccaf4b0689cdbd1eca042a49 : SUCCESS in 1h 56m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bd58f5b191594cdca91e18d3d0229764 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":1},{"id":"50d5ce8e8d8eb5b7be5df004d63ffb759becdb94","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-04 22:09:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"07b40aeaa24f74fb64fc8a46121f23f8aef4f165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 22:09:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebe8b65ff783ce614645eef4f7af2453e3f3a22d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 02:28:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f11a6d3911f4d2d8ebaff1ef2c5b215 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/186b133c546e41a9bd8546ecaa89ce18 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9425d092852a45f0a965b84eaa430ab8 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf9fd2cc4b464e4294d64c051951a3e5 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4fd4316d84f42739bb7caeff05f5356 : SUCCESS in 5m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6616bfa2f0241e79aa71c83678912c3 : FAILURE in 13m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/debb9128d5f6493fb39a1c491336528f : SUCCESS in 1h 37m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/27e8f8817e3a4bb58c96508aa36d6379 : FAILURE in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3a74e0ccca1b16cc606212f8fa400de96a59a07e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-05 02:34:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3ac83703b8347822d465ae074e6fa6c731800ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 04:23:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7dc4de48af6a45b3b39a3524bb0345a9 : FAILURE in 3m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e18e24e136646b69b243b513a30e6a4 : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/815b0c1af89f46ebad7118797f785ebb : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0666594614644e75bff53fe7eceee830 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23a8df3eae544082b9bf89ec531d382f : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b3eece1145b4f67a37e9772fb15ebda : SUCCESS in 3m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8dbe368cf1b74f7d9e9ea77953cccf59 : SUCCESS in 1h 10m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/51fcccaf0b2a47b7b5d856f0f0bd8658 : SUCCESS in 1h 43m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8fc46941eb484459b8249e8ca1838862 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":1},{"id":"967c151813b388d7c4f2d3d5651623397494c709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 04:23:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4244f07eca3ad9af31bc6d7c71e5e20f29dac95d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 06:53:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"99698dc0f3f6f1773ad09bbd73535606a2824110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 06:53:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1eb2417a3a4425954830f9acdb4028 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/041f472382584753b4e5f001dd166967 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/105484e537a34a7b929480ce473c6bdd : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c679b71d7b843f398943f5fe81dfcc8 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35917b89754b4775b95620a154b90a2a : SUCCESS in 3m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3653ca97b54e482a9d15f13f0e23ad96 : SUCCESS in 1h 05m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1346c6ee94fb463c9a7051c00cd23e8a : SUCCESS in 1h 36m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5ae9aa8669d944f2962830fbd9175dca : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1e6a4340258ab30f4d16f80bd2c63ef784441b89","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 21:46:42.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 56810b0c1fc070d5aee0187e0a637bff00ee4a09","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83c946d3ded550fe2cac5672915195825d1cc312","revisions":{"83c946d3ded550fe2cac5672915195825d1cc312":{"kind":"REWORK","_number":1,"created":"2020-03-04 16:42:55.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/71/711271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/711271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/711271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/711271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/711271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/711271/1"}}},"commit":{"parents":[{"commit":"6b1b94be547ac00d53fb80671bd15a1245db3807","subject":"update constraint for oslo.utils to new release 3.41.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b1b94be547ac00d53fb80671bd15a1245db3807"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-11-13 04:20:42.000000000","tz":660},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-03-04 16:41:47.000000000","tz":60},"subject":"Requirements-check: use local repo like any other library","message":"Requirements-check: use local repo like any other library\n\nThis ensures the playbook has done a local install of the repo so the\nscript run can assume that it is installed and can be loaded without\nadditional handling.\n\nConflicts:\n\tplaybooks/files/project-requirements-change.py\n\nThis commit also adds the change to use pip3 since the CI runs\nthe scripts with Python 3.\n\nChange-Id: Iaf759b183e6821bf6690195586e7e65bf0533c82\n(cherry picked from commit af94ab84790e73a86333c4c6fa778827851d720c)\n(cherry picked from commit 1c270c8dd53f2aa37d93e21d353f885682cdcf08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83c946d3ded550fe2cac5672915195825d1cc312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83c946d3ded550fe2cac5672915195825d1cc312"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
