)]}'
{"id":"openstack%2Frequirements~711297","triplet_id":"openstack%2Frequirements~stable%2Fstein~I259975af909ceca2b289eae0b76c4fc4a779b7e6","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I259975af909ceca2b289eae0b76c4fc4a779b7e6","subject":"update constraint for oslo.utils to new release 3.40.6","status":"MERGED","created":"2020-03-04 18:07:53.000000000","updated":"2020-03-05 04:44:45.000000000","submitted":"2020-03-05 04:44:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711297-1583383485960-4545ba28","meta_rev_id":"1cdc618d1754b71a01d9fcc73293d1a599791f6d","_number":711297,"virtual_id_number":711297,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-05 04:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-04 20:00:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-05 00:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-05 00:57:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 20:00:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-05 00:57:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-05 04:44:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e16c3d3f008a574fabbc27aab8cc09b94e8597aa","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-04 18:07:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da464f3ba6fa6597497e48434a503fc628c5beef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-04 20:00:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e4dff486ea71d24afd4bd19f67b3dc17c1703c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 23:40:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c7eab35fabe64e0d905c29477f53ba22 : FAILURE in 5m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776d329edf964a9f8b1a6820e109b54b : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57a96bc5beae4a8ab0037e5c46089e0e : SUCCESS in 4m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0378d9bbaa2845eb8ff0ceca3f605ca6 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7a8f6b7271b4391852f0e10e64c2e9b : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8ffc6abe8914a999684372e622cd327 : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a15cf26932a4e54bfd4537e8218583e : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8fe5d75dd21942249cbfe41202bcc854 : SUCCESS in 8m 56s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/da706c21fb3c4f9b95f1fb84a01ca30c : SUCCESS in 8m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ddcf8b54323d4e2db00146f2bda26fc1 : SUCCESS in 7m 58s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/75bdd6b213c4400186848b6b97fb8533 : SUCCESS in 10m 49s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3c6f657ad9da441a9f2fad52f6673d7d : SUCCESS in 11m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5c29f108bfd44295baaef811ada186e1 : SUCCESS in 6m 00s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/ed63f9729c5b4aeb97649eabda64ebe3 : SUCCESS in 6m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c12638b1e1b34883aef0bbc06ff122a9 : SUCCESS in 10m 45s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/05619c5142af4780bd314fbc649521a4 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa95801932854439b32e8b1334e6f0b5 : SUCCESS in 8m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e5e0f77e6a864455b3886f882d6727cf : SUCCESS in 24m 12s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/1aa576b55aae4dceb0f857a367ccf476 : SUCCESS in 30m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/38c5002a19c744deba8974511cd35b05 : SUCCESS in 25m 30s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/12dd3df4c4c84123b34b0d103b4034b6 : SUCCESS in 26m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e4e498bdc8e44dfbbc96359d8c50dfb1 : SUCCESS in 14m 22s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2ce505659cb546959d7f34fa060bf3b4 : SUCCESS in 15m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ca8bd2f54794215b1de94a33988aa5e : SUCCESS in 18m 56s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/b75012d75efc4f73b45d863e270aa716 : SUCCESS in 8m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5d1789983eaa4ef0b2df217043f94b02 : SUCCESS in 10m 26s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a872a5839f4747b58a95579367ca403b : SUCCESS in 11m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9ca1b3b39fe457eb2e5197677ab190a : SUCCESS in 1h 07m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/df866132c6e644f887ac4660f5a38932 : SUCCESS in 1h 45m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ce7bc1946f374d60aa7e9a3e898134db : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0936f215583215cc8ec0f63b1ff373fa16fe3773","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-05 00:57:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf8fbd3f140f0ba202fe94c5d3176834472b9cdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 00:58:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef4d23c4df7aea24ec73a1099211fbded3ebc3d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 04:44:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/203fcc1bdd8244f1bff444b8d2d58bb0 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3245f247716a4dd8b78642abb717701d : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d157fbc810bd41dfb311fee3b2c011ba : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3331c7e128b543faabace3dcc972af5f : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d6c2f9759c94c3b85d3551c893b6d19 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/172098d1f60148b8846eef03763acbd7 : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6d7acabb6b024409857c90c5e98bfd0a : SUCCESS in 8m 01s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f53a615adffa4beba4c7fea6867def5e : SUCCESS in 8m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d94f678fef4c4456a70b890b7c2fd3a8 : SUCCESS in 8m 45s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/89ea0cebde19491a8d60059a00934c61 : SUCCESS in 9m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/558d0ef68f394c41b965de323593381a : SUCCESS in 5m 21s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/36c4555e603d49dbb5a5f8560d0a69d6 : SUCCESS in 4m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a6ad04f8e9e646659238421c3085ccbe : SUCCESS in 11m 35s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/519be19635d84fa9a1423269bd3073e6 : SUCCESS in 12m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/283d0685e7df4d98b69382c2128264ae : SUCCESS in 6m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2681dffe579e4684aeba247dcd45bc8c : SUCCESS in 22m 09s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e8e28ed1ffa5472798c5d7f588dc5090 : SUCCESS in 27m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b0143a86b213440296092b61dd54f6d9 : SUCCESS in 24m 16s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/b18803280e8f41449a065f0a5575af81 : SUCCESS in 24m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/72ed43b72125407eaa9b590c5f249c2c : SUCCESS in 12m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1353498e274f42348418f8c5c6355357 : SUCCESS in 12m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2fe33c6118d49879dd359939a257e73 : SUCCESS in 16m 50s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f12204b259b84e5e92c45a5c5f7bcb4e : SUCCESS in 5m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e8dfed1ad9ce46b6861bab4bd367fcec : SUCCESS in 8m 30s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/6325dd2ed19e4059985b45a9fcba5356 : SUCCESS in 8m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f89118e5c52f49e28d6a029e2269cc70 : SUCCESS in 1h 11m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8a6463e592e248bcbefc74e10678858c : SUCCESS in 1h 25m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/abbd1eca053b4efc85ee40a3aae3b3f6 : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6b70ca011013cc5caef2f2aa374a87ab8d25acec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 04:44:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e4bfce958d65f6050d428698c2a2545cba7d8d2","revisions":{"1e4bfce958d65f6050d428698c2a2545cba7d8d2":{"kind":"REWORK","_number":1,"created":"2020-03-04 18:07:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/711297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/711297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/711297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/711297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/711297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/711297/1"}}},"commit":{"parents":[{"commit":"e796f0333c6ed32dc6184e3d7413ffd1ac66f91e","subject":"update constraint for neutron to new release 14.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e796f0333c6ed32dc6184e3d7413ffd1ac66f91e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-04 18:07:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-04 18:07:51.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.40.6","message":"update constraint for oslo.utils to new release 3.40.6\n\nChange-Id: I259975af909ceca2b289eae0b76c4fc4a779b7e6\nmeta:version: 3.40.6\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Ie91e964db15d39d253cb5ea6b475ef744ec5283e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e4bfce958d65f6050d428698c2a2545cba7d8d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e4bfce958d65f6050d428698c2a2545cba7d8d2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
