)]}'
{"id":"openstack%2Frequirements~711841","triplet_id":"openstack%2Frequirements~master~Ie8354566da3ffa1b9cac15fc66c7b39314272103","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ie8354566da3ffa1b9cac15fc66c7b39314272103","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-03-08 06:25:06.000000000","updated":"2020-03-11 00:57:19.000000000","submitted":"2020-03-11 00:55:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711841-1583888129488-eb552905","meta_rev_id":"ad3e124cafb659d9c5509376d6a78d10661b4ee3","_number":711841,"virtual_id_number":711841,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 00:55:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-09 15:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-09 15:51:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 15:51:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-11 00:55:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba7bcb05a149afcfe8b55a48a22f3c1ad81d7bb3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-08 06:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78f83e751c5de3349d834062921e3ec83411e91d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 08:21:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c029efd470140888e1f11d4405ca271 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a5040b623bc4f0c822f414567b393ed : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c9777d744954553b8b02604862030fc : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05e00559d37f4651a11ef7b7ba4da3c0 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d57c0d2d4fb4f24b2a9eb45188251be : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92fa0f7ad4d74e3a825bd2920510d522 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e6f5b9450f644e5b80f41e98b2de3eb : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4dd596ff60fc43fd960e0697633b4207 : FAILURE in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/26e9b6f28f9d497391b553310387a8f0 : SUCCESS in 10m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ef34a2e7b8f34a7aaff516618a6d87c8 : SUCCESS in 4m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cdbb39a4f54043d4974be19f2c08681e : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ec8c59bbc7a4e98824c1a60fe3f3266 : SUCCESS in 7m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/116f0fe4c26c480fbfc6f4d08281d800 : SUCCESS in 13m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee97a58ccd7d4dccacc3a1e54d8ba261 : SUCCESS in 32m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b0416336cfa44a2b8d739a881360130c : FAILURE in 12m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7b6ac76364ee456ab3e6af92edba7289 : FAILURE in 13m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/968022e54b81402290a2307ccdc27efe : FAILURE in 17m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/561c140d889e4b7389ab5d5b87adb34c : SUCCESS in 5m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/83a1210d80c045d3a2a5358edb5e49e4 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6548b172d9440aa92602ceaf769e1ef : SUCCESS in 7m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/32f0b7bbce1e49d6804679b35cf676c6 : SUCCESS in 12m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4530d1290e4749049af21e9756b14393 : SUCCESS in 9m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be236db90d1d4f9ba95fe0dfa93a00f1 : FAILURE in 29m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/acbc0b89181c483ba9fb945846902d30 : FAILURE in 52m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4b8cfa09abe4a78aafa8cc909d53668 : FAILURE in 24m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/62a2545a27f142b89c280f84a8a1c11c : FAILURE in 49m 56s","accounts_in_message":[],"_revision_number":1},{"id":"698b980e85fc4d64cc06a291a28fdd0c71121c85","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-09 06:18:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dba2d91fc9b16a409da6240d459f4aeddef82148","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-09 06:23:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6d99a168d4897e3179b54cb727ebc5e6649d8eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-09 06:23:05.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"eb4e6061aa1b513be99d427b2d11aec50db469fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 08:10:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d31f46fc91944bd9b06a6b44113859ee : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56280bfda99409ba2023973e1038fd1 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e880839fa9fb48fd9b51eb2703abb8b3 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8deefdc3f02543d693ae2a7e54f0c008 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cadcd7a59e8f4d9b8049f3859bdb2499 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eef457efc8f0479f828e0371d38c7fbf : SUCCESS in 7m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f652017f60bc43f692eb2818e8d4dcef : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b520a2eea620438284679bc339d83692 : SUCCESS in 21m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9734db0e931f4b48963c26ffdfa234f6 : SUCCESS in 13m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/69d17d3b30ea4f94b381b385992454ae : SUCCESS in 8m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aa9038f3bc5e45e0be1c00a19fb202e5 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cc5d5a89297432b8bdce9594aed2e04 : SUCCESS in 8m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8b54c2c16a914cf39c13ab07f05026a0 : SUCCESS in 20m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f519f2064724d0bb55e2b253d14ea39 : SUCCESS in 23m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/461a07551f434d769d5cf2a9d68045f0 : SUCCESS in 13m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d6c324a6b0ab4a0bac010e157b8cfde5 : SUCCESS in 13m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a44cf36a84e94baea3b1ab08baf81f97 : SUCCESS in 16m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/38848fdf076b4ba5af86ebefd2575186 : SUCCESS in 5m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/845e5f12514d4c279919188c8360872e : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66dca2dc7028452cadddab011f032aec : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7ae952b878634e3488fd6eabb99d7e05 : SUCCESS in 11m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2608129210454de9a6ba149ba0b29be1 : SUCCESS in 11m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd529ef90d5c44bd8efed72bfc756717 : SUCCESS in 56m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/732fd9d6f21f44979a188de2c8541f9a : SUCCESS in 1h 12m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3ae77f01e1549f29cf66410f4af994c : FAILURE in 1h 28m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/804d0d626a904d1bac7e86734ea9ef97 : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":3},{"id":"49156f18bd0ef378a6429b46942dc608d329dc2b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-09 13:59:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1491d15ac174fe49bea2163b7ff3f92f8217480c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:32:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f436ae802f4e9eba7195fe45828542 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2e8e94bcdc49ceadd96145146cf77a : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/259defe01eda487ea03f3801f4aa6b74 : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f14f261d465467f84ada827fec454fa : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbb67404492646db8464b9c4689003bc : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9625e38a425c40a1b8e2075d0ccc2ff4 : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9507a205a0d4734a7a61e52cf70dcda : SUCCESS in 3m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/60626905a6e345bbbd94cb0cd080516b : SUCCESS in 10m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/316faa29b86f4bdba8170fc3b77af338 : SUCCESS in 10m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fc13b996106140158758ebb5331e2781 : SUCCESS in 7m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6fbcfb9665a24694af6a2b70f5804728 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f8f54ea70a54f7f9f30bbfe7ab53219 : SUCCESS in 12m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/39f4bfd273284e3a89584b0ff71c477c : SUCCESS in 16m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6314fff932b546fbb774b17797f9fdda : SUCCESS in 24m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/68ae95ffcd294861bf98059734109c1e : SUCCESS in 15m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cccb98f231454854b315eb1002cd3b95 : SUCCESS in 15m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d59e8d6aff024e699cb131dfd641ba31 : SUCCESS in 16m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c9a91c3b769a48a68f2900126dc3f171 : SUCCESS in 6m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6abe4e8278ee4d1d8c304ee69edde9ea : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7bd83c6af93042e19f611c8cb9df0e1a : SUCCESS in 6m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/34d7a1cb67584b29a619766ea1a56472 : SUCCESS in 11m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/61fda726245549eaa44bd323416d7166 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bb921c82bc049d7bddcace2b34af98b : SUCCESS in 55m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c0a19e2fb134bc8a162571c3a31287b : SUCCESS in 1h 04m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d230e499de948eea0c0e18caf17fa24 : SUCCESS in 1h 26m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f1775d260624d83af505d97893b8b74 : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":3},{"id":"da876a9425de25c74552867e496f544058724d57","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-09 15:51:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fba865bd757cd70034b9c23c1d0de9a161980a88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:51:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c06c8f6f482c9f8ce3c8eccd067b69170525bd3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 18:17:41.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0653d7bd11404eb29df2a22b811d0760 : POST_FAILURE in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d2bc402565469791e6cc73e7595ba0 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65d6ce570e3d474b82ead773e08e1f5c : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef8b2f2d1d034154a0f8cacd1cd5344e : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94bec03e0ad141068316c27f0d104565 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a75dc6df28f64275993558ae78837656 : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b80aca5d7de4c5cba1b563cadb2a14d : SUCCESS in 9m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/56ddb7ed54854e849ae602ab82d97941 : SUCCESS in 10m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/030d812a60ec418ea39286286d330d9a : SUCCESS in 5m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fa2b5d639f5d4d9dbb511c6fd317537d : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88461731bbf647228e93a2abbe0389fe : SUCCESS in 8m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/94a571050fd34f67b8cab87680398d2b : SUCCESS in 19m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7a9b3d15a2fb4b728bc067b15ba82102 : SUCCESS in 29m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b842e780bf7f4e8e9dc827c81ce2d158 : SUCCESS in 16m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ac6efe3c40a249eaa8d22e765a1eaeb9 : SUCCESS in 14m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e5d935b6dd24cfaaf9b6444212c97c2 : SUCCESS in 18m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/61b7ce1cfa2d4015a8073f7688825ee5 : SUCCESS in 6m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/838a2e1794d149d3bdb0c9999bcd2984 : SUCCESS in 7m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/717fbaab577a461c8447cd5789ef493a : SUCCESS in 9m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6b964346264c49c8a267663638e15da1 : SUCCESS in 17m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/63228c6157e8401c8e10a88ed350dd13 : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/620541506a984177b4910b12ad17c601 : SUCCESS in 55m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b1a17cfdcbf4da29af8283127bcf585 : SUCCESS in 1h 14m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6cb0e37a5f174b358bc72aa55c4d3ae4 : SUCCESS in 1h 40m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/67b57c9bd4044a178a2070eda5c55ca4 : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":3},{"id":"d45d3d6c3778a99b202ed36d9ef858a573d82206","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-09 19:54:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bdfb482e28b99bc2202155407158b0e7e1f03a5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 21:24:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489ff45bc89644b8a1dd6b781ea0975c : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9879f464ecbb440d89acb7fa681431ff : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5483b549586d4a579c04ba8fc918fb6d : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/727f4fab0f5243d692b6309b85c229d5 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd61c71920844142b218a66728e85b06 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e36e590fb8aa4f4f8dfc636376c96581 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f5fcd7460994ab6b5b370a213914feb : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c5b39ee661994d6c9f53fd36b599a8a9 : SUCCESS in 7m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fe81ddc0f894445681940b4d70eca755 : SUCCESS in 9m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1ebc2d21798f459cb098bb1705fb5004 : SUCCESS in 4m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/209e9b58dc014cd7a9aab7e592b082a9 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27c43e761bfc4c5685874f152620cb1a : SUCCESS in 8m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/213fb1752d864dd79f23bd5075ebd1ec : SUCCESS in 17m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4486df948fc248e6b9b8e809f8c96869 : SUCCESS in 21m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b68f6c3650074183ad9fe7f989d459be : SUCCESS in 12m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/66ffaa17ea0d4ad7b1add451fec500aa : SUCCESS in 12m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3acf665c3ffc4cd294019a1840c8e5cf : SUCCESS in 15m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4998f974e221455f81a496345682168b : SUCCESS in 4m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a6c5340171ba4835aad39c6f7eb9d7a8 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bfa7228dfc7746148c8e809f0fa0ccc1 : SUCCESS in 6m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/33b2bda93a5a4767a872d58644fdf8c3 : SUCCESS in 10m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4da00fd2db8648a4bd82fd5d5f1bbf47 : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d497ba9d59824c9f9b4a43aa065f891c : FAILURE in 54m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0543e4c2f6324ba38db57e81c8099d95 : SUCCESS in 1h 01m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ccfa033587841edbdb88dca56d7a027 : SUCCESS in 1h 25m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a0c1e8e49c22499d8b7d04d82017aed8 : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":3},{"id":"f02215cf0544f85cb169820b8eee08a23dcc7c9c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-09 21:27:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a6c1ae0a4210eaa00b2c14fc711cd2ad257f493d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 22:53:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cad07b52fc2424790a8845058e3b4fc : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f2b2c026554519ad2f55c10cebed81 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed4328c6256f4746aca568d24943571f : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/914cde0be6a44401b221055607299b64 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee74a554f9764e359964b80fb647c6fa : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2105d47badd34ad58e1be6ce4a5eaeb4 : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac1514ba94354064a322d0369601019a : SUCCESS in 3m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/82d3991d8c3946919e6b3d41cb592598 : SUCCESS in 11m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b72d33628ad74f218874c15e5ba21441 : SUCCESS in 8m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6c26c2a5eb8943caa315f2333ad2a908 : SUCCESS in 6m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1f6bdb67441f49489224d20b725baccb : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4245142e27484a308d61708336114d02 : SUCCESS in 9m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dac30585560340f786537c095055e668 : SUCCESS in 12m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/636c02af50634de1b4d99af5e24d4079 : SUCCESS in 28m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/83e8d3de783e4f49b35207499d1ab1ec : SUCCESS in 12m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/371d4895dda64be09af4660cafca8b06 : SUCCESS in 11m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3fb94ff3fc5c4495a7c5118c742d6d09 : SUCCESS in 16m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1bfaf922e9904247b433e6ad9541eb57 : SUCCESS in 4m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/14f5427f2837463ca7ffd1bffc6467e9 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b7d4a154d06486b83871cfd36732357 : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7331f09bd0514a4da10caf731f5bd100 : SUCCESS in 9m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a05e9e8cc06543f38b0eba69639382ff : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdea61467c1b47a6b4859b98d5afc0e9 : FAILURE in 55m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7296fc5ed0114e86aea0f3f7503e8a23 : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c58bede8f55483a89256ca63a3ec700 : SUCCESS in 1h 21m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ba301c1bcc994773a2945f1c811829fa : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":3},{"id":"a2dc10e13f01df33fd0bfae11d4df837eaa82327","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-09 23:04:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7480998c1a0f205a4ab36847e422dcc9f049f763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 00:38:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5ea4cadc2b455e929358faa9fe2d55 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8decff0add4d4706aac37a818ad740e0 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a06376218b0e49079bd464f862e77087 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/730cdf57ad9945b08e1912f5484fbd68 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac2400227368417fa5e0f481d27fa2f6 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1809720e785e4b10bea4726e8744d688 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/310a3d21462b469e890b6d1074ef2b3a : SUCCESS in 3m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/997097920b4345c88f329a5614b57ce1 : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e142b4cdff6e4d81be6f13a8624196b0 : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/45ee9abcd0914b428aad610ffb014f41 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/89f070f0864b4d37bb03da68ab547b5b : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd3f48d4599547648c8a884c48ef7c6d : SUCCESS in 8m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3acd197d03754baab91208d79a38d749 : SUCCESS in 18m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6e4f8a7298b04945be79ca706c075c07 : SUCCESS in 33m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/24777fcbee84488b9581e59d57bd8c7d : SUCCESS in 24m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2a2f8360964b45c7ad9ec51f22655e06 : SUCCESS in 14m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89be8daa93164ceda548cc582403c91f : SUCCESS in 25m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/37efa3bbf78244038c47145ecdac910e : SUCCESS in 4m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/64b1ee34770f4a799102fbe3dcb89344 : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64b35fd196fb48b8bf19a80506ea725b : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cb86549aba2f468991ad95058be6b332 : SUCCESS in 10m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/16bf6120021c4d1393521806d469c992 : SUCCESS in 18m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/421f5162872f4dd3b2d183c95535e5e8 : FAILURE in 1h 04m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae664db646aa49ea95aa0e99e4618375 : FAILURE in 1h 26m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d20deed623544503ac4ed08e1a3c73b9 : SUCCESS in 1h 27m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8f83ac2fc837436cb90fd733f87dffd0 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":3},{"id":"e75a7470c5e5678bd779726df1b1b41af261e74a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-10 05:02:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"96465532748821caedeb92f36027a975f19b2d57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 06:18:57.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e86f7c3161144c5baae4254887f5928 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23be4a5f4754eafa53151214485f29d : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5eb49398c5b43019b2b7cce3949dab8 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7687fad3c309499da93d8881850ab1aa : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb218d9de71a4ca9981c11f47e62dfc3 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/857656f5c75f4e10b64c0eb68df00570 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61e02bcfc0c148119acd28c401a3ae6a : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1f9635d88b0f436a99105509d20735d0 : SUCCESS in 7m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/95182e1f157f450094a9945d9c90f99d : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b46f358b5a924d929f0b8d1529f84568 : SUCCESS in 4m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b49009941d6245dab8fbb2e83c9b2ed1 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e97481d5ac549c384d433c63a661728 : SUCCESS in 8m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3a2e62719af640c19c52fd3ddbcc4b08 : SUCCESS in 17m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8ab496928a8747939c102dee228c190b : SUCCESS in 23m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/97a50f9ced3c4da686d7d3da94675a36 : SUCCESS in 11m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c97912b6afff401f8bd691e73d8f869a : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d443d185fe94a359023b0eb65882efe : SUCCESS in 16m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d7100f5bec304bfb8b8ec9fda366bdd2 : SUCCESS in 4m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/23740c53bff1423aab0e894ca2bd1190 : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/75f4c13361e940f3afa574013c944f19 : SUCCESS in 6m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/46d540d2729b4a3bb3e09f12cf225c5b : SUCCESS in 8m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7c3aff83e1aa4f6dba314f4b0641bf2d : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6ef90281e17431d960d128449031cbb : FAILURE in 57m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a758ffcd63e84cf1aa451ab6e1f1fba9 : SUCCESS in 57m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06fbd5a3a0d8443a983e275204ee84ad : SUCCESS in 1h 14m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c93c28a24d6c4da1adbdcec19d6fe1fa : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":3},{"id":"e1895a26dd9aa2dd36628a92c8894089740a0a6b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-10 18:13:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"09f257bfcd1768cf7d237ea8c3110ddf32af92c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 20:07:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f3b147cf0643fab70e9a5bf350c7eb : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6673b01869f24ba0b08115064db9e8aa : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bb238e8885946479e492a7999159f88 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3e1c9cdb97247b4b1d76ad0825736b3 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/490b806dc08b44d7b3f33e4210f00bbf : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/63b010b6fb4a42ab9b0a2b422050f10b : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b27834d7628f424bad9eea555f62ce0b : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/32d03dd8702d4e19b46e338441f6c9ac : SUCCESS in 12m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4ffc839e256a47388a9c7e134b775569 : SUCCESS in 12m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cdc757bd77e3480680ac60e29b2b9c38 : SUCCESS in 6m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0de445995c6941fa95384bbb35bfd86d : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6723e39f25947348c536cd6e8b56c4a : SUCCESS in 11m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/01a28e037c2b440fbd16d5b67ae15283 : SUCCESS in 20m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a69a2880eb3b4ea0a3efccfc0d2f2471 : SUCCESS in 25m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e2d3ee467d0048468f2b6bf5b530201d : SUCCESS in 16m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e146745a984849fc85bb6e342ab0cd43 : SUCCESS in 16m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df4e1dea9b4f43078e3c46dc729f2c4d : SUCCESS in 16m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0e97c1edd4b1430495ebeb887603ca12 : SUCCESS in 5m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e5541f8011594b9e911ade78284b8347 : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe8cd68bcc374846a58702938ac40da8 : SUCCESS in 10m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1140787d284840b481eeb8c1a86807de : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2b686c5e84534332a7c398f5d3d1ad16 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6327f9ba367542b19a6c7f58f498ad7a : SUCCESS in 53m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a0fc26005e3476eb19c3b20c73f4fe7 : SUCCESS in 1h 09m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/caaedb0dbeec4e2ea5a9e43b846518ad : SUCCESS in 1h 32m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aa20e57157114e7b850ca97e56c23aff : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":3},{"id":"90e9a1eac4b2208c63c1ea495d8524ade8cb1bd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 20:08:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"371d9e3bb0d774dd8405f3b2d3769bf980328419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 21:54:46.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daa53cd70e83447585037e6922eaabd3 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/570907a8c2554053949752c6312f5825 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea1abf0dd7fc4840a2fac92d8b6b5514 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e326997407d45f3b4be955e70e015f8 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f114aaad7ae415981382baacf7419b9 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff47de2319c0455ea2f8bb5aca13039c : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/22dbbb3b174a4d65bb32712bd2a83a30 : SUCCESS in 10m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b21b936f95f14e5fb2dd07bfb556f908 : SUCCESS in 13m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b521b08087234481a66f88eefee87a09 : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c3eb2686fc104f2c90ae45e3bc8fdabd : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6591541a0b3a456bac85d46c96a9ff46 : SUCCESS in 7m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2178763d68994f3bbdb5a22eee87b891 : SUCCESS in 23m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d59bf1800914baf9fd1e7a8381da36f : SUCCESS in 28m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/990214d3cdc544819c506901c5815d23 : SUCCESS in 14m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5faef94f173e4135b0d1ed14597eae5e : SUCCESS in 14m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1e7734a86f64cfea83fa9c9b78da9e2 : SUCCESS in 21m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/65b531b6aa7f4392ab6f8417c2d1d066 : SUCCESS in 4m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b02f66b98b9a480f8f1604f8f5a6f763 : SUCCESS in 8m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3583a3162594a7083c12badd53ab578 : SUCCESS in 7m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6651c8c7363d4b2c8285ac54c2511736 : SUCCESS in 9m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/eb4ece1246034631a8b62f7f955efff9 : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb9f800452bf411ebf12019765ffc57c : SUCCESS in 56m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b62b50bde3e944689fd96873853a1f7f : FAILURE in 1h 19m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eaaba2f84840416bbd40f69b526fddd6 : SUCCESS in 1h 28m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/40eccf2f39ed4927a4591efad2be3776 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":3},{"id":"79b835f2fccd0ac492e0e77bb59dec61d0fd6282","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-10 22:00:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"684402d49f5943bbe0a7e73a2dc6ad33cda7035f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 23:27:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c1dc4979a94925b63bf07fc521d98b : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c78d4d0bea4ae68188c0e704af158b : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d56240917dc437eb25484ade005869d : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/360cd4a7ee4f4cd599e8e11a5aafbe36 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/957780827b364ae4802820484c63c1b1 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9829e58bdb7b43599cc871f1841bcbe7 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5378d3a9b9e14cc5bafe4149408f576b : SUCCESS in 3m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/923d4f72e0924962abc25f28e424ddc5 : SUCCESS in 7m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a2161ef279cb4bcca4046a1cc9610c68 : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/582f938616ad43a591937c6385cc5cb3 : SUCCESS in 4m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1e9d37e0cd8d4528a6c4f9dbc7b2777f : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bfa715bc7a844fd497d3ea36f6af8c4e : SUCCESS in 6m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ac3c28b0c1f74e08bcd43636a1389b51 : SUCCESS in 14m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9c2fd485af854b0dbe44242e25051d9a : SUCCESS in 24m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a60d7df195f54dd0a0cbea3407e89f1a : SUCCESS in 15m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b82e9d8c72534ff8916677f884267ca5 : SUCCESS in 16m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ff8ba34bad54f0ea3fc0fc4bb361d2d : SUCCESS in 17m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1897af6135d74383a6ca0cc0e6b3928d : SUCCESS in 6m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b877da734918487bbd4e4175a749a024 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e29237f0ef11485ead7b775b09e15922 : SUCCESS in 6m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9bf438f9595c43079a84541be3f9afc1 : SUCCESS in 8m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/09bccad77111485ca3495abdc7452051 : SUCCESS in 9m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dde4772c978d4ddf82e8cea92091e95d : SUCCESS in 56m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0450370472754158865e7a564891e0bb : SUCCESS in 1h 05m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df3a50c3d4be48ebad878f06c746af82 : SUCCESS in 1h 21m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8b0a127b9a674e308347260b89132701 : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":3},{"id":"00589a2d067ba5f9c3b602bdcd3ce75c503b7221","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 23:28:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"03e019dba5b72ca9054ea80aa67f4ffdc311b02d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 00:55:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb0946ab46c4e06b94060d5b176b7e6 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4078c221b77a41c1977564488302539a : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/048e9010577d46c091d301868e6ead5e : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eedb8dd8ae1a462cbafe8caf832ba35c : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d620b9048f345cd867c312b4ca447d6 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ab2859f21dd4522901dd2babb08c77c : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/36423aa196df4be6aea48d57a8690b45 : SUCCESS in 7m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cb38cb3b3a5d46ecaab35e6271107610 : SUCCESS in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dd47f79a98004cfe88c8c9469af9d757 : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8c8efb3f4bf942fabe21a56d6df7b303 : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c162fefef811482991fdfcdc27b693c3 : SUCCESS in 7m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fec48af2e2f641d786e7056b94cec14f : SUCCESS in 13m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a19f38553e044cfd8d70102815a1f591 : SUCCESS in 22m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6dd8bc85c93b4c429250b730fb642df3 : SUCCESS in 13m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/137356abf9b94d099d317c6055460aa5 : SUCCESS in 12m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40b5de49afde4a3db654ce2d97e55aa4 : SUCCESS in 16m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bc03cd28a666427aae85477dc15a1165 : SUCCESS in 4m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/32e671d9c6294326aa777ea3131d6b67 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a148d03740c4d27abd805650c583d47 : SUCCESS in 7m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3d946760038e46489131082aa8728b5f : SUCCESS in 8m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b4089d6510ff4dfba320fba2ce334b36 : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20a1d9a3842b4863a234727d01d3c618 : SUCCESS in 49m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51a795d357d74595881f45c8aec89f27 : SUCCESS in 1h 06m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3aea6772cfa4fa592a6a5c757fca3b9 : SUCCESS in 1h 21m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5f207597222c40518860cb7868c5ed7c : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":3},{"id":"99bc7402ef3a979295a0cc8a6a893737712b2c73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 00:55:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ad3e124cafb659d9c5509376d6a78d10661b4ee3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 00:57:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417b2ec1e93b48a0b99144c7c73df169 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"abe9c1b3ade21144e73e113ab466b67ed418c710","revisions":{"c102673017b4ab14593f1ee5b066875daa28ab6e":{"kind":"REWORK","_number":1,"created":"2020-03-08 06:25:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/711841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/711841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/711841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/711841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/711841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/711841/1"}}},"commit":{"parents":[{"commit":"2f2c92a3a3bc53eb945d42166c11ded359f5db33","subject":"Merge \"Updated from generate-constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2c92a3a3bc53eb945d42166c11ded359f5db33"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-08 06:25:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-08 06:25:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie8354566da3ffa1b9cac15fc66c7b39314272103\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c102673017b4ab14593f1ee5b066875daa28ab6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c102673017b4ab14593f1ee5b066875daa28ab6e"}]},"branch":"refs/heads/master"},"dc2ebb6b04ac49c0ea7fcf4a340278e464766c31":{"kind":"REWORK","_number":2,"created":"2020-03-09 06:18:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/711841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/711841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/711841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/711841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/711841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/711841/2"}}},"commit":{"parents":[{"commit":"2f2c92a3a3bc53eb945d42166c11ded359f5db33","subject":"Merge \"Updated from generate-constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2c92a3a3bc53eb945d42166c11ded359f5db33"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-09 06:18:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-09 06:18:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie8354566da3ffa1b9cac15fc66c7b39314272103\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc2ebb6b04ac49c0ea7fcf4a340278e464766c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc2ebb6b04ac49c0ea7fcf4a340278e464766c31"}]},"branch":"refs/heads/master"},"abe9c1b3ade21144e73e113ab466b67ed418c710":{"kind":"REWORK","_number":3,"created":"2020-03-09 06:23:02.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/41/711841/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/711841/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/711841/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/711841/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/711841/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/711841/3"}}},"commit":{"parents":[{"commit":"2f2c92a3a3bc53eb945d42166c11ded359f5db33","subject":"Merge \"Updated from generate-constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2c92a3a3bc53eb945d42166c11ded359f5db33"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-09 06:18:13.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-03-09 06:22:02.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nhold back docutils because boto likes to cap things\nhold back msgpack because salt likes to cap things\nhold back alembic because nova fails tests\n\nChange-Id: Ie8354566da3ffa1b9cac15fc66c7b39314272103\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abe9c1b3ade21144e73e113ab466b67ed418c710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abe9c1b3ade21144e73e113ab466b67ed418c710"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
