)]}'
{"id":"openstack%2Frequirements~712098","triplet_id":"openstack%2Frequirements~master~I749a92ee79119208a3e96ed9769391003ac105f2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I749a92ee79119208a3e96ed9769391003ac105f2","subject":"update constraint for neutron-lib to new release 2.2.0","status":"MERGED","created":"2020-03-10 14:29:57.000000000","updated":"2020-03-11 00:30:03.000000000","submitted":"2020-03-11 00:28:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712098-1583886505657-cb471cbf","meta_rev_id":"d424e306e1e2c817800d95d80dd88aa1336710cd","_number":712098,"virtual_id_number":712098,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-03-10 19:36:32.000000000","_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 00:28:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-10 22:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-10 22:00:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-10 19:36:32.000000000","updated_by":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"reviewer":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"state":"REVIEWER"},{"updated":"2020-03-10 22:00:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-11 00:28:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fbd321a6a30478f131c176e6c0b2379af094963","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-10 14:29:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee201d39fd4465b5682a5f7ca8e6ac55cc19acd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 16:06:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/746c7b0b983343d4a79528ba9f279a53 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acad9eddb3ae4da8a3edef7151f07aca : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0613f13dd5f4312a26be6519c9a979b : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acc6e7df84e74539a994a90b4e6ccbcf : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00930ef544594e9fa246a6efda362b33 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/abc7d43342cf405cbb6bbb17cfe1bb85 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f47972300bcf489cab9f91bbfcb6b3c1 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/802b6c7aef9440adb2743235f59b1e08 : SUCCESS in 9m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c3a85d27166f4004b8990875ff7c1c22 : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1533945cb3b848aba5201b75544a4346 : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3aef1d0b63bd43c6bc842f9a9bea3c7a : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a5a938b107e446bb94ed48b16c21d74 : SUCCESS in 8m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ce5afff8a9de479cb4813b1659706a7d : SUCCESS in 19m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/39e5f6ad42094161a0033c8f57d3faa2 : SUCCESS in 22m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/91fd1bd7b0d94776bcc05af942dd6527 : SUCCESS in 15m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/236d37e579c84566a0157e933900900b : SUCCESS in 16m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4ba26790e724c688d7ca0d33a69e174 : SUCCESS in 19m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ffbaa3dfd66d4c7bb26e405a9bf7401e : SUCCESS in 5m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0d70d242c49f4a279897875e1b7137e6 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6084c680505d407c904ab321e56059f0 : SUCCESS in 8m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/094c4c1bfedc419f9f1c6672cc89f665 : SUCCESS in 10m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/af36e4236b3a4680a1384d8d069f22d4 : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d337a98b8b5c49d4a9927b6a3b9e5726 : FAILURE in 50m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38a0913a62d94cccb84d01ec4dc2f9e6 : SUCCESS in 1h 02m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5ea5c032c954619887d9593ad3ab3a8 : SUCCESS in 1h 24m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b128bf310a324295a3d5d90e7dd1cd39 : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8c454e02b0f16656e0313c0a0e415254728492eb","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-10 19:36:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c3c2840c4be44d63ffdafcacfd7a2eefd7a5c71e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-10 20:44:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f35fe214a9a13e596803f1b5da326f19cdab7991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 21:12:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b137225641854f5f8006a66678cee22f : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302920398e9140688717c2e9b0294a9c : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/530cbcae52444dcfad818924036e06c5 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c332a3cec6954111b46d531d8d652ec3 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed2bc4b90c1746769d8b83feadec1a20 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df5377e8345f421da2f40d2cf90c5b7d : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f512405eaf8f4347a73189aa4f883192 : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/83d631e3203749c69e1f5dfe08d169eb : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2350762337ac407aaa6be1197309be9a : SUCCESS in 9m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3d3d430ead4f4d779dd6415c2c733ad3 : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/142f7c5324b1446ebba846aaf9ca90bf : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03deb4eaa5e34406b4f477cb83fbed94 : SUCCESS in 9m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/42dd693caf6c4f90b1241078ab316bc8 : SUCCESS in 19m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b220e74005bc4b39abe25a55660daf41 : SUCCESS in 26m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/23a64dc862ab42e08b6656e5e78d0550 : SUCCESS in 16m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2199b1b4663049488ec50433c7c8c425 : SUCCESS in 23m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/60a448f0e4a64d0092b3ce691fe7df72 : SUCCESS in 18m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/43347ac540614562a83bcb4cd5683131 : SUCCESS in 6m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e5ec1ba9f2214bc29924f173d7ad1121 : SUCCESS in 7m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6dabd9fdcb69452697aa8ad23845ddee : SUCCESS in 8m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/abbdb5ef8ae04630b64a39eede451745 : SUCCESS in 11m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b38dfee6f3d424eb16f2871f528c457 : SUCCESS in 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4fc78f2afb546df87f39682a96d990e : SUCCESS in 58m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/199f7b43020943eaa404290d0c60e332 : SUCCESS in 1h 06m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/656cdc9fc6b74badbbf5fdde478709f5 : SUCCESS in 1h 23m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/26aad38912e44bd1af9ae2d08ecfe11d : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1abed4576dbdefb4425c97d7684d970cf7475e8d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-10 22:00:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0cc4fde9bd700c0c0d7bafda7e90dc0cbaf62c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 22:01:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc261fd73f204e46392d30f38fcd45c278f7a8b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 00:28:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"889eba268cdde9e66084235635c87b2fd4207011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 00:28:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db971927ac2c4e4189bfa71c79b188a6 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52a31c7019948fb8211e783b4f196a9 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7492fae8ef2d45b195df817acf10323b : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c03b69af98d842ce9afe2b45f2690a68 : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4024ed79272e4687b1391666a88fdcd0 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3534682bcb844f2bbbf213150aebb755 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f0bd1cdf1fe4bb8b26feeb482533fb5 : SUCCESS in 7m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e419f503d95f4ec3a2dedc1a7e57376d : SUCCESS in 8m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/28cb84602f8d4de1b45b9a9b27f39acd : SUCCESS in 5m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/def6d43fbef24d8c884ecea7993ff2bf : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4f593d96cef4c4e97ccc8f8377e7c93 : SUCCESS in 8m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53d7cd2870a046b892d633433082da4d : SUCCESS in 12m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/57e200c54d10421fa98f07d94e43aa34 : SUCCESS in 21m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6aa0e45dcb1e49e0b1dddd6a30404542 : SUCCESS in 12m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/edea7be9a4af4490b0c8c67799a4a96c : SUCCESS in 13m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b0810f687964c57bec19e3fa5100aef : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2fb7c1379d8b4ceeae35476936d527b1 : SUCCESS in 5m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d0e9a4fcd2f64f359fd3b861cd3e3a45 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/98634642ef424644970785df4167b1f5 : SUCCESS in 7m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d77e8d72a5db4784adb893718e723989 : SUCCESS in 9m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b40e0112a8e2477ca389b80ff9cf435e : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/752c816965f04a409a0a1fd49f4fdb34 : SUCCESS in 1h 00m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8ebd96538e64847a87d494fc9bf8f29 : SUCCESS in 1h 01m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b019d4b937e242f98404c7f60cc8df89 : SUCCESS in 1h 12m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f6ec45ced54b45109bd8a5bdeff72041 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ed8d1aa48acefd431a8196a85c3eb6d55f025e48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 00:30:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb317fc2aaeb41c9a6217baa342efe80 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12f8421dae14a5c45de5d2a66fba427db16441b9","revisions":{"12f8421dae14a5c45de5d2a66fba427db16441b9":{"kind":"REWORK","_number":1,"created":"2020-03-10 14:29:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/712098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/712098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/712098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/712098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/712098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/712098/1"}}},"commit":{"parents":[{"commit":"2f2c92a3a3bc53eb945d42166c11ded359f5db33","subject":"Merge \"Updated from generate-constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2c92a3a3bc53eb945d42166c11ded359f5db33"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-10 14:29:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-10 14:29:56.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.2.0","message":"update constraint for neutron-lib to new release 2.2.0\n\nChange-Id: I749a92ee79119208a3e96ed9769391003ac105f2\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Change-Id: I95d99f9aa1eb850ab8d6d2c1223d92fa8b58dea9\nmeta:release:Code-Review+1: Igor Malinovskiy \u003cu.glide@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12f8421dae14a5c45de5d2a66fba427db16441b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12f8421dae14a5c45de5d2a66fba427db16441b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
