)]}'
{"id":"openstack%2Frequirements~712288","triplet_id":"openstack%2Frequirements~master~I56c3da84ec88f99260ec99517385fc016c48fffc","project":"openstack/requirements","branch":"master","topic":"1866961","hashtags":[],"change_id":"I56c3da84ec88f99260ec99517385fc016c48fffc","subject":"Setuptools removed Features feature, fix last version, fixes #1866961","status":"ABANDONED","created":"2020-03-11 08:56:27.000000000","updated":"2020-03-11 17:07:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4c5a8a487d510f449ed34d439731f00328d98ec5","_number":712288,"virtual_id_number":712288,"owner":{"_account_id":30238,"name":"George Melikov","email":"mail@gmelikov.ru","username":"gmelikov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-11 10:49:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-03-11 15:30:52.000000000","_account_id":30238,"name":"George Melikov","email":"mail@gmelikov.ru","username":"gmelikov"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2020-03-11 15:27:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30238,"name":"George Melikov","email":"mail@gmelikov.ru","username":"gmelikov"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30238,"name":"George Melikov","email":"mail@gmelikov.ru","username":"gmelikov"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30238,"name":"George Melikov","email":"mail@gmelikov.ru","username":"gmelikov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-11 10:49:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-11 15:27:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1ed70dbb8eee6d6730d77f923c006b664f6b573e","author":{"_account_id":30238,"name":"George Melikov","email":"mail@gmelikov.ru","username":"gmelikov"},"date":"2020-03-11 08:56:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01a80b2f3bdb37bef3f9a81dfab8854fba53cb6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 10:49:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a789997e60d5409ab6bd0d00ddcda5e3 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4babf5be02c4862b835f5fef1e4e503 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a3e500e2ee64788b76ab31b7fe9c5d0 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69c85f13abb949e09926d279ee7a8d11 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51f71aa0215a49158b43f384397123d7 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89242ed079944f9ea20839945165a14b : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c9b652a990143f6bcac4374de663287 : SUCCESS in 5m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/29e0913c72a64613a3f0e42467a799f4 : SUCCESS in 8m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b4af15abd85e453481201e8cda363a07 : SUCCESS in 9m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cfe5408d4383475db83993201fc09568 : SUCCESS in 7m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f3a97aa35b0c496089053c2ffa7b0e9a : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/085d1e21805b43b19594fc7c6e96fd7c : SUCCESS in 11m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/556bbff400bc48979436b87ee30f21cc : SUCCESS in 18m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6202bcfd30d14b6d92a502465d50bc88 : SUCCESS in 22m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2572c40a636646f587c79d5847b9a8cd : SUCCESS in 17m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/aacfbedbd9484342b3a03009c3034aca : SUCCESS in 13m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f440c9a2e2324b8284d185fd45a71496 : SUCCESS in 20m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1fe613a2306842e5b3d2eaa294abea74 : SUCCESS in 5m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/08c2d9a8f5724741818f7111b06f3a5e : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41dd8a2094d74642945b95f7bc81aeb7 : SUCCESS in 8m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c6e764b755f9435985912eea8d0c2ec9 : FAILURE in 6m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/eb91565967e64976956647757c48b6dd : SUCCESS in 13m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13c60ee6abe949aa9f12dd110b41afa8 : SUCCESS in 53m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38e1f406ec634e388ae77051b2394330 : SUCCESS in 1h 07m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c01aad489d844f1a59920ab8b20397c : SUCCESS in 1h 48m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e309387bcd16420c9b5e37f24f6e3090 : SUCCESS in 1h 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c0af56d86cc150113b577bde8ed7146bc48683d9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-11 15:27:30.000000000","message":"Patch Set 1: Code-Review-1\n\nI don\u0027t think this actually solves anything, the reason we don\u0027t track setuptools is because openstack gate installs it BEFORE using requirements for anything, sadly.","accounts_in_message":[],"_revision_number":1},{"id":"0f6a50935c734b132248249746dbc15c8b8f4c1f","author":{"_account_id":30238,"name":"George Melikov","email":"mail@gmelikov.ru","username":"gmelikov"},"date":"2020-03-11 15:30:52.000000000","message":"Patch Set 1:\n\n\u003e I don\u0027t think this actually solves anything, the reason we don\u0027t\n \u003e track setuptools is because openstack gate installs it BEFORE using\n \u003e requirements for anything, sadly.\n\nYou\u0027re right, thanks for review! \n\nI was too late, looks like superseded by https://review.opendev.org/#/c/711930/","accounts_in_message":[],"_revision_number":1},{"id":"654d8372c2973b30f3907799b4d45f5789a53aac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-11 15:59:44.000000000","message":"Patch Set 1:\n\nyep, mind abandoming this?  I can do so if needed","accounts_in_message":[],"_revision_number":1},{"id":"8e045918e759de790729a16bdc38fba6bb9d9188","author":{"_account_id":30238,"name":"George Melikov","email":"mail@gmelikov.ru","username":"gmelikov"},"date":"2020-03-11 17:07:18.000000000","message":"Abandoned\n\nSuperseded.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed0731c96e7dfe84d1a772929164c1d281424e43","revisions":{"ed0731c96e7dfe84d1a772929164c1d281424e43":{"kind":"REWORK","_number":1,"created":"2020-03-11 08:56:27.000000000","uploader":{"_account_id":30238,"name":"George Melikov","email":"mail@gmelikov.ru","username":"gmelikov"},"ref":"refs/changes/88/712288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/712288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/712288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/712288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/712288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/712288/1"}}},"commit":{"parents":[{"commit":"3b9ce573c7802dc258de373ac655a7d3318c7b0f","subject":"Merge \"update constraint for openstacksdk to new release 0.43.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b9ce573c7802dc258de373ac655a7d3318c7b0f"}]}],"author":{"name":"George Melikov","email":"mail@gmelikov.ru","date":"2020-03-11 08:46:51.000000000","tz":180},"committer":{"name":"George Melikov","email":"mail@gmelikov.ru","date":"2020-03-11 08:56:15.000000000","tz":180},"subject":"Setuptools removed Features feature, fix last version, fixes #1866961","message":"Setuptools removed Features feature, fix last version, fixes #1866961\n\nBug: Issue 1866961\nChange-Id: I56c3da84ec88f99260ec99517385fc016c48fffc\nSigned-off-by: George Melikov \u003cmail@gmelikov.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed0731c96e7dfe84d1a772929164c1d281424e43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed0731c96e7dfe84d1a772929164c1d281424e43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
