)]}'
{"id":"openstack%2Frequirements~712653","triplet_id":"openstack%2Frequirements~master~I9fa112d6867dd6a4e599837ad2640f233bd13f16","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9fa112d6867dd6a4e599837ad2640f233bd13f16","subject":"update constraint for ironic-lib to new release 4.1.0","status":"MERGED","created":"2020-03-12 11:06:53.000000000","updated":"2020-03-12 21:36:33.000000000","submitted":"2020-03-12 21:34:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712653-1584048882540-3bf77f9f","meta_rev_id":"9f838be13155731afabd9b23d0e5bec06779d8b0","_number":712653,"virtual_id_number":712653,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-12 21:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-12 15:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-12 15:37:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-12 15:37:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-12 21:34:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a378d226faf18e231aacf39a00004fc17b211ad5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-12 11:06:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c7d1d59b516ba940f23347c6785818af04e8eb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 12:47:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42713e6c8c9d4ac3bbe4aa3944333215 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36d38df6a5b449f9d81f005a0b448cc : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/043f4c588b934480a1f89636d0855f64 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cbaeb174e26457bb01e15a1a3911b57 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd4da00937614455be481ee6d8b8c2be : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/603e91c7453040cd869a6b62346526ca : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51a583c5f8cb4c628f6afa08216314ef : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3d39116a269648e09f61fc7780b46263 : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aaa8c290d3cb4bb18a2eb1bacd9baa32 : SUCCESS in 9m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4f1245f5483e455083b4a6538282ec19 : SUCCESS in 4m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/91390637533e49bc877bb08e9827e09a : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eda2591ac8db41368b4bdbb7f79ac660 : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ae7b63e75ee64b6996450fccff27c42c : SUCCESS in 16m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/30868918c0c646a4ad813ee6359efc0d : SUCCESS in 26m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/060307eeed3d491983b34d64dbc9725d : SUCCESS in 13m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7ad4d2d3ac574e6eaafcaba02fa49042 : SUCCESS in 13m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a2a68b926f04fdf99e645468f571040 : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8f1e7c25c1f94fc985a3369ccc56c0dd : SUCCESS in 4m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4ab32cdfd7284bca82aba9b29e965a7d : SUCCESS in 8m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7926c1e5c8354031b2fa6d12726af118 : SUCCESS in 7m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6e329160c14b49429c5c37c7c98d848e : SUCCESS in 12m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a34da85b1fbf49698aed12b2ca446a90 : SUCCESS in 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f7cbce38a6548b9890ccedabb52b3e0 : SUCCESS in 52m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1201bacf5d5c40e89f9d0891bbf93513 : SUCCESS in 1h 12m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8046959dfdc46f9b08d9417b61b10e9 : SUCCESS in 1h 25m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/79f933d3a30e4b2f8493cc8886d3a16d : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":1},{"id":"46397cadbdc72dad212329ccbb3f7dfd65cf7e53","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-12 15:37:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7f63a964f66773e7005c1dad62269a8bcf362e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 15:38:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6adfb55407263a6932066443da5df0fb8cf1753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 17:16:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eefb996c23546998922a31f4d5ef5f4 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a93fec804146a4a537f42bbf7c9a32 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a4528e8766642cfaffbdbab67ba0f07 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa460811b44a4c5c862be40d66da31a4 : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa786ef1a4c14e089d6b196e38029205 : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f8befb8df1d4a6bbbe73f58f964fe20 : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/53a9ae4ddac647158b18add6490eed82 : SUCCESS in 13m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/41796f8e32a8422ab27ea981d1837507 : SUCCESS in 13m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e198cbebfad546a7a4344ff871677b41 : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dae9c10e10464e0ea701f9dd76d684a5 : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f106ffa7aaf4c788bebf343e4972b4e : SUCCESS in 8m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1368a2d6132f454e8dbd6dc049bf30fe : SUCCESS in 16m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c9be8cb887d44f92a85998f223e328f7 : SUCCESS in 28m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/94e627f05e7a419d94ae35790620fccd : SUCCESS in 25m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/13fe2da93db0446199c199249a4f3e22 : SUCCESS in 14m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4e887495d9746d49d4885c6a83e6ad2 : SUCCESS in 19m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2cc3031ee46c4c3086d022b6f372a84e : SUCCESS in 6m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/df3d53fd12dd466fb7662368f175b147 : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9681e0eeb33148408454bcdf36208eb0 : SUCCESS in 8m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b31395518ada4dc7b4055b50d3a0cd7a : SUCCESS in 13m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cbec158d902d42baafc24e0a32072c17 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e8549d029784d45931152774f23bec6 : SUCCESS in 56m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1bf8bbb8b8224c71b00a06402fe70507 : POST_FAILURE in 18m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/973162223fce4224b9bb0bb87fa1c258 : SUCCESS in 1h 18m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c8d58c4f65094ef0ae537e03ad3b74e2 : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":1},{"id":"dd8b7cb6f63243e0afc311711a80fa7b12634526","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-12 17:18:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aea9724a217afc4b31bf84ed3d936ed84add6dbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:58:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1226fd6694ef44e6bca2f5a9824f4241 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36ac0eee286e4a16990fe75665efe6d6 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce2b90536a9b40a981c819ac3e0d869d : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c5908a0f5e8435f8065c69e1eacf075 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/407b1b4b871e4dd182b4c708d2e79edd : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb9a6201d87a4552ba11297a19d7248d : SUCCESS in 5m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/acd274d692c14eec8f61715aaf6e261d : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6b2ebe2cdbef4efcb795c4a84e5889ae : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b13129cdd154c6d9cf09667ea6eded9 : SUCCESS in 12m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/438c05b7ae4845809ddac15046f6747c : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/482b090e25154ea9840b303150501767 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36467181ff4f4026a6b66429a83ef28c : SUCCESS in 9m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ebe894696f704e0dabd4181d81fad2f5 : SUCCESS in 15m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/caebaf519bdf42e3b2e5ca7c4443299d : SUCCESS in 29m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/654dba58dee942e88dc48c10630c4107 : SUCCESS in 14m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/876698c21d9046baa7be415dd1f89e96 : SUCCESS in 17m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d84438dfe8a48e696d0af9a51371806 : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bf22bfa5ab8d449084d29f7b5e6c79a9 : SUCCESS in 6m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c299d2cb82344ce4b719e216d1a14580 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2a9db24d7794f3ba74ed3090f7fce2a : SUCCESS in 7m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1167b305b9604206b91792b5383da93e : SUCCESS in 11m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/37a0ae039bb54bdfa42b89378cac1815 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e47ba8184f845c4b949cbc7ed2ec240 : SUCCESS in 53m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/008dad5493f144c98b09e56da811a5ad : SUCCESS in 1h 03m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66c849f1f7f64ba5bb6599a7c6908e6a : SUCCESS in 1h 52m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/73ae022a421046e9a9a65d7f604d3d47 : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1fd628eceeac6df73a6895e6a26e18508f7f52cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:58:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33b420d7ae33d46fefd24645e127b5f8d30d21bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 21:34:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d92465df739204056d028696f5ec3793bf72ba1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 21:34:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd629dea5af449ff91a58b91cbbe01c2 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bff46c7184574e09bd37ea77d41872fc : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/060ac5b0ab684d0e9e9b651f863f085a : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77b794b607f14f9b82857ef730140886 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd538e8ae9e443a78446e6620be99144 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5b44cea1dae40a9b7ef281bb4332f80 : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1444c740e49e4773afbe1a99d9c596cb : SUCCESS in 8m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/66aee91043b5416ea2f9fc076c6ab361 : SUCCESS in 8m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ed660fc4a3fa41e1a938a268f396c315 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ea8386bdae1e485394b7b8f1634dd706 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/594feaa2d7324990a8827a0ae64b2588 : SUCCESS in 9m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bdc11fa6bfb142cba18718d543046b63 : SUCCESS in 18m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d94c7ecbab8748aaaef5dc7c8be470b9 : SUCCESS in 29m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/619f292be2e640b795f2ba5fd2fcb746 : SUCCESS in 13m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e729393f614a4ccb88b1b4a81dbc5878 : SUCCESS in 17m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0669ac26ef0147fb9f36f4efef01ba59 : SUCCESS in 17m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d83122bd4bbd4ac8adc5b9ee7dbee370 : SUCCESS in 5m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8d3bca405b5243fc890f96b2cb333792 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52a742457a774e7581031b61817f400d : SUCCESS in 6m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bd1a08da823143f39cf5a00cd7a07d21 : SUCCESS in 11m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d8a8aec800864913ab8a871df8f21273 : SUCCESS in 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01ea24cbcd2e4ba180b8612b46ba9c1b : SUCCESS in 52m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d92caaf567b24e2d8e0b66693272ea74 : SUCCESS in 1h 02m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2fc2e991f8a4d9e93547fdec8a77eed : SUCCESS in 1h 31m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a02573e6a3804da982bc45632ce995e9 : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":1},{"id":"774584e0d4e571dbfd1a9c2e335f80b0cc03063c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 21:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8e3a22bc5047a287a9ef248b6be721 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c0d8f331fef735f86fba985dd0c7650cc243b25","revisions":{"4c0d8f331fef735f86fba985dd0c7650cc243b25":{"kind":"REWORK","_number":1,"created":"2020-03-12 11:06:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/712653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/712653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/712653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/712653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/712653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/712653/1"}}},"commit":{"parents":[{"commit":"cbb853ea0dde696ce89480d5f95cd53d4ee2dd51","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbb853ea0dde696ce89480d5f95cd53d4ee2dd51"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-12 11:06:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-12 11:06:50.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 4.1.0","message":"update constraint for ironic-lib to new release 4.1.0\n\nChange-Id: I9fa112d6867dd6a4e599837ad2640f233bd13f16\nmeta:version: 4.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Commit: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Change-Id: Idf5988f9bec9c70ff039ac7462c88601b15782aa\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c0d8f331fef735f86fba985dd0c7650cc243b25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c0d8f331fef735f86fba985dd0c7650cc243b25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
