)]}'
{"id":"openstack%2Frequirements~712942","triplet_id":"openstack%2Frequirements~master~Id63f94e708f027555ce5194d2e76da4c8dd2ec42","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Id63f94e708f027555ce5194d2e76da4c8dd2ec42","subject":"Fixed pytest-html constraints","status":"ABANDONED","created":"2020-03-13 13:08:45.000000000","updated":"2020-03-14 11:31:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"276059f3ca4dd1fa049bc03cfeea0363f3261b8a","_number":712942,"virtual_id_number":712942,"owner":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-03-13 15:37:49.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-03-13 15:32:43.000000000","_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-13 14:59:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2020-03-13 15:44:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":1,"date":"2020-03-13 13:17:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 13:17:47.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2020-03-13 14:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-13 15:37:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-13 15:44:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1ba9ab9ba0016845e68584b1ec8068e0e30aa33b","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-13 13:08:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa358704e1d552375bdcf0c7d457a8a1569426e9","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2020-03-13 13:17:47.000000000","message":"Patch Set 1: Code-Review+1\n\nThis is breaking all jobs areound taht are still using python27 (centos7 especially)","accounts_in_message":[],"_revision_number":1},{"id":"babde286a2f5cc477e30ad5b80d96e7489f3a737","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-13 14:52:48.000000000","message":"Patch Set 1:\n\nWhy do we still have py2 jobs?","accounts_in_message":[],"_revision_number":1},{"id":"1d434dedb87b5ce24983a6c646a44ba846b0a5a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 14:59:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6fe89c2f52491f91a56a9e09ed021e : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d02a391789044669b197462cafd66930 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e61869b69c6b4c33b0c22768c7414fd3 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20cf599144094ad28b0d77d5758a661b : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07ccf423df4e4045af5f6d14ece43c96 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/043c7c851dee436392c4e93ec8bef8de : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4311ad79ec88424eb4dcb3bb444ff100 : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a03f8fa0b9a400babdc969270d9fe03 : SUCCESS in 7m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7e413bbc0f194602a5bb37efad022e63 : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ef793e04d30a46478884666122b2909f : SUCCESS in 5m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7b28cbe005134557a8d823109d1d8fc8 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd909206ab014284a138ad392c390aff : SUCCESS in 6m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5497b2ad05864774a69622689083bba0 : SUCCESS in 12m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9b782f55463a48ee84c0605938879f19 : SUCCESS in 24m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/718516d165fa42219a4fb06a89237811 : SUCCESS in 11m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/27b7240b11f54ca9a08bb41db1ea698a : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc0320680b7f4956bb97ab6a3a6c24f6 : SUCCESS in 16m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2d38e723dc0646fdbc0c99fa45ffb6ad : SUCCESS in 5m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bb858e2eac404397b178f60ba531ab92 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e87e6c7356e4dd8b015e856964cb52c : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/117231d51755463b8679c6860615c5fd : SUCCESS in 9m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/109bb369106a4ba6bfca1d616893dc74 : SUCCESS in 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08a9f2ceb61d426ea526367e95c26a5d : SUCCESS in 56m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7c02cbd8dd64719996a44fdf3191a56 : SUCCESS in 1h 31m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9e1fe135d884613adcf21151deb3bd3 : SUCCESS in 1h 21m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cbbabca4f6704e2f85f27c112605611f : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f37dc73f481262c4db304e82b3d68556b41ce46f","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-13 15:32:43.000000000","message":"Patch Set 1:\n\nWe still have testing components running on py27 (not shipped).\n\nConsidering that swift still uses py27, and that 30% of the upper-constraints file still has conditions for py27, this change makes sense.\n\n\nWe are in the middle of the transition to py36 but this broke our testing jobs.\n\nMainly fixing it should give us few more days to finish migration and hopefully not to care about py27 any more.","accounts_in_message":[],"_revision_number":1},{"id":"8be4347d6e0e67b1701c00e07d72905534aef4c2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-13 15:37:49.000000000","message":"Patch Set 1:\n\n\u003e We still have testing components running on py27 (not shipped).\n \u003e \n \u003e Considering that swift still uses py27, and that 30% of the\n \u003e upper-constraints file still has conditions for py27, this change\n \u003e makes sense.\n \u003e \n \u003e \n \u003e We are in the middle of the transition to py36 but this broke our\n \u003e testing jobs.\n \u003e \n \u003e Mainly fixing it should give us few more days to finish migration\n \u003e and hopefully not to care about py27 any more.\n\nI don\u0027t see this component being used by swift. That should be the only community project still running py27 tests at this point.\n\nWe can add this now, but it will be removed in the very near future once we start cleaning out the unnecessary entries.","accounts_in_message":[],"_revision_number":1},{"id":"9e75f8cb34e7d967daeb9ce9f17e2adcb94eaa5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-13 15:44:51.000000000","message":"Patch Set 1: Code-Review-1\n\nnot only that, but I think that this will be reverted in the next global-reqs bot based update to upper-constraints as this is a diff against a generated file.  As a stopgap, ok though","accounts_in_message":[],"_revision_number":1},{"id":"5f746d5c7570929ab5d2e649cadcf16c0fa7da60","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-14 11:19:25.000000000","message":"Patch Set 1:\n\nlooks like it was included with the bot fix that just merged https://review.opendev.org/713085","accounts_in_message":[],"_revision_number":1},{"id":"276059f3ca4dd1fa049bc03cfeea0363f3261b8a","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-14 11:31:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e49d8338e210272bdc9fedabb60e52e76b269d4","revisions":{"3e49d8338e210272bdc9fedabb60e52e76b269d4":{"kind":"REWORK","_number":1,"created":"2020-03-13 13:08:45.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/42/712942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/712942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/712942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/712942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/712942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/712942/1"}}},"commit":{"parents":[{"commit":"df7163634ae9916a6a6003b649ee183dc1331f03","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df7163634ae9916a6a6003b649ee183dc1331f03"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-03-13 13:08:35.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-03-13 13:08:44.000000000","tz":0},"subject":"Fixed pytest-html constraints","message":"Fixed pytest-html constraints\n\nPrevious value was making py27 installation impossible because\n2.x dropped support for py27.\n\nChange-Id: Id63f94e708f027555ce5194d2e76da4c8dd2ec42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e49d8338e210272bdc9fedabb60e52e76b269d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e49d8338e210272bdc9fedabb60e52e76b269d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
