)]}'
{"id":"openstack%2Frequirements~712976","triplet_id":"openstack%2Frequirements~master~I78b3c7f8fc6bb0759b8a4e382ad18f43622f5b97","project":"openstack/requirements","branch":"master","topic":"fix-pyscss","hashtags":[],"change_id":"I78b3c7f8fc6bb0759b8a4e382ad18f43622f5b97","subject":"Use pyScss2 and django-pyscss2","status":"MERGED","created":"2020-03-13 15:35:17.000000000","updated":"2020-03-17 20:35:38.000000000","submitted":"2020-03-17 20:30:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712976-1584477023074-7e8b974d","meta_rev_id":"00e6d1e073a9b5e70956d779fdcccf47628601be","_number":712976,"virtual_id_number":712976,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"date":"2020-03-16 12:39:07.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-17 20:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-14 20:05:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-03-17 14:26:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-16 07:16:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-03-16 06:53:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-03-16 13:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-03-13 20:30:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-03-17 14:26:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 20:30:10.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2020-03-14 20:05:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-16 06:53:45.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-03-16 07:16:26.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-03-16 13:07:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-03-17 14:26:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-17 20:30:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12f2cb16e86c16ef800d62e62d4ff2d8f0378b0e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 15:35:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"493be0f6e21b6aec3178ab4ccece49f9c2fbc81a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 15:49:57.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f3dbce5ff171b95fe53d0b30dc1461ea2856392f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-13 15:50:51.000000000","message":"Patch Set 1: Code-Review-1\n\n1. thanks for doing this\n2. we need to have both around for horizon to be able to pass gate while it\u0027s switching from one to the other.\n3. does pyscss2 use the same namespace as pyscss?","accounts_in_message":[],"_revision_number":1},{"id":"5097f0315c65551e38d62857e4fa473fa603f1be","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 16:06:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3b0241b5ca3f96aeaae881285a0ef233463cd73a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-13 16:13:33.000000000","message":"Patch Set 2: Code-Review-1\n\nI still think pyscss needs to stay in requirements until you switch horizon to pyscss2 (at that point we can remove pyscss)","accounts_in_message":[],"_revision_number":2},{"id":"4549a8268196a989267b82ac27fa95203365fc7a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 16:29:28.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"112df8b7e7815fa2689d614771a9e08a6be36067","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 16:31:26.000000000","message":"Patch Set 2:\n\nThanks for the quick feedback, Matthew! I\u0027ll resolve your comments once horizon Ci will be happy enough.\n\nNew libraries have exactly the same API so we don\u0027t need to change any code","accounts_in_message":[],"_revision_number":2},{"id":"810707f8134678e8a7fd126fbacfa2c50f229d04","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-13 16:33:14.000000000","message":"Patch Set 2:\n\nIt\u0027s more a question as I see this fork is now in eOne\u0027s repo:\nhttps://github.com/e0ne/pyScss\n\nWouldn\u0027t be a better solution to fork under openstack namespace, like with ryu / os-ken ?","accounts_in_message":[],"_revision_number":2},{"id":"f9c7faa077474f26bacb0a27c70aa6a4570e135e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 16:36:31.000000000","message":"Patch Set 2:\n\nLajos, the reason why this code is in my own repo is that I just wanted to unblock gates asap. I tried to summarize all issues in the mail[1]. I\u0027m OK to move it under the OpenStack governance in the future.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-March/013322.html.","accounts_in_message":[],"_revision_number":2},{"id":"01c1a2b64015293c5c180348adb66ec9d07a15dc","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 16:56:32.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"14ab1b5df9ba9d9a3f3696d68926ae0186bd8b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 18:10:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/413f8158e3164af69eec8d142b4e32a7 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1fc5bd536f645babeca9a10b557183c : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f6695118dbd481698bc73514d4d24c9 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/432fa82ccc3f4a5c9d0cc67c53a15266 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20494d38b4994a75913db52c4f4ebed8 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec494657c63a41a89c78c5a54df484ad : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7519149242d74b48aee4b76481898fbb : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d11d797e29764106a785b74a00053a93 : SUCCESS in 7m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1b051f8b299c41b397549698f6f35a0b : SUCCESS in 11m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2df5c65d2064413aab864729e777a729 : SUCCESS in 7m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7c198a3ac2f848af865b767be4303481 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bab600f04f24482c88819500f5961640 : SUCCESS in 10m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e3c36c2529543b7856c48178e0be543 : SUCCESS in 16m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c03f8d4bde1b4a4d82daacfb7c3ce104 : SUCCESS in 29m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ff988a99ab34414b9daa4437e2f5d3f2 : SUCCESS in 17m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2dca2d8d638b46deba3afc79df8bd8f8 : SUCCESS in 18m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0bac9416de624382aa9adb881b80366c : SUCCESS in 20m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dc653b46761846e68177b7fda903f689 : SUCCESS in 5m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a3e9fcf6dc894c908848b93ac5829094 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/27960866ce6e4fb78e9f98e175c63955 : SUCCESS in 9m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/31172904b73a410b9f8ec9e60eddbe72 : SUCCESS in 12m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/056e226b5bd44ebaa0e378bb77366f6b : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbad05b57ea944c78627aac41d8d384a : SUCCESS in 1h 03m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81301cc8e61644c08f6fb6d1d1d2b6fa : FAILURE in 1h 02m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06fad7dc5eb9430e89dcd18bdc1d2c11 : SUCCESS in 1h 42m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/268e6b2b2dcc4710b53bfb0f0c6321a4 : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8dbeab697ee635e818e60e2e5c9dd257d50e1771","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 19:07:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4db2b6294b715bc56cf0ff295ab222937cab63d9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-13 20:30:10.000000000","message":"Patch Set 3: Code-Review+1\n\nThank for quick solution, as I am also waiting for this to be fixed in my reviews.","accounts_in_message":[],"_revision_number":3},{"id":"c82a6339f7e48bd3da872d64ce9e857e938d0444","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-13 20:35:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"112195a8bd624df0689b04bebb53727abb90b4de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 21:07:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c472428d007243da8d73e1d98fd44199 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8795ad1365eb4d5d9737ad1bf92f1d15 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1d7f7bff9734f8aab8fa825fcb0b1b5 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7bea72075fd94e87b198926d396c595d : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8da7b1a4820c4729a96d1d1bcb82d86b : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/746d28bcc4a04dbd956ddab3d6e0b421 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc5bf18bdfb54b428c3e254c7722fcc2 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0b8c4f0d5ce3449495c8db8c949b28ae : SUCCESS in 7m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a9682794d7b4de8b0053ac9742dc3e6 : SUCCESS in 12m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1fb817c4ffbf4281abba820e77673f46 : SUCCESS in 4m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5b8c35022918493faa599b4e51ab1ccd : SUCCESS in 13m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10695a2009cc44efa295abe810c347cd : SUCCESS in 7m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f419b4b115f4cc680b311b55169d7e1 : SUCCESS in 29m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c787ed738ad04c0c84e8a1e03d43cd3d : SUCCESS in 25m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b3801c85ce74838bbae5606c657d062 : SUCCESS in 19m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/664521161fcb497c8abb7c9f7a7cf2a5 : SUCCESS in 13m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/752ae4a6c78c46e6933706b1c919747c : SUCCESS in 15m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9525326866ac41ad83a90a47350be1b6 : SUCCESS in 4m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3c531b74afaa4ee7a1dd4133397c463c : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09884d2c1bef4643b92722e180205dea : SUCCESS in 7m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/71b1ff4530844d869606dcb691fad774 : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bd95bf3da05d481894e104c80ae83acf : SUCCESS in 9m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40bf6cfa50b5427488038285ffad2c13 : SUCCESS in 1h 26m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d71b049398bb43e2a5f1529c36f82592 : SUCCESS in 1h 12m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b792382e17e407183baeb0541dd2ccb : SUCCESS in 1h 55m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bc66253783ff44c4babcc1a5e6e7ed4b : FAILURE in 50m 10s","accounts_in_message":[],"_revision_number":3},{"id":"c4a685e2109860823fce212f4a54e5d2631c44c0","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 21:09:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5deca91da24a819c7abe2678a13a668d2602d167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:38:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275096f99c6440e187d4eeb054bdfb87 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4028d4c120740ae88cfada03238717f : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/559ff7d1251c41899a70905da1f1001a : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f79d66589e74035b8ca15caf90091f3 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1264be8cc6db48f7a0c2cbd860f7e752 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d933c14dad294e7ba64245b11813a285 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e39a4a7b6a774731af22b0cb832dd2b7 : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/31bb1c1a1d9f4350b83ff810254e4541 : SUCCESS in 7m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7ee84c901cac45b99bf3d5f0cdfb9768 : SUCCESS in 8m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4f446150d5d5451bace0a24e46d745aa : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fd2137bd40704bf795e08ea88eab2c76 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ddb450b46dc4322a5df9822b4de348f : SUCCESS in 7m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/547a69de09ba49829532f4d52ee746bd : SUCCESS in 21m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f2d2e90246f845d899370bacdae23889 : SUCCESS in 22m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/081a2c25c5d34c5cb955bc41aaf9bd7e : SUCCESS in 14m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1c81bf03a67c41bc8fc22aa0c723fecf : SUCCESS in 12m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2290564e599142089dce7c1e87d9d866 : SUCCESS in 18m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2ce8a50ff8564e96929bdd2bfac56edf : SUCCESS in 4m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e55ec6226ed040dba6a53461757cb387 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ec62a33a1b143c794e356a374f4d696 : SUCCESS in 6m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1a0a4966330f49a3a26855a61d1d739a : SUCCESS in 12m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f7d6595b81294ed087045216f0842d95 : SUCCESS in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca9502ecdb0c433091fafe0c1c177032 : SUCCESS in 1h 27m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/10caaf3275a3483fab48d16f79de8352 : SUCCESS in 1h 01m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbf3cfd6bd4943c0946089e87e98c1a6 : SUCCESS in 1h 16m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a54022d5e59d4f30a48a1957975a3863 : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":3},{"id":"d820a9d46a1a06376191e1ad51b039370f237b6d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-14 16:14:07.000000000","message":"Patch Set 3: Code-Review-1\n\nif we merge this patch, all projects that currently use pyscss will fail","accounts_in_message":[],"_revision_number":3},{"id":"81c68568016d1cd6aa78311142f0ee193461534b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-14 20:03:13.000000000","message":"Patch Set 3:\n\nMatthew,  this patch adds a new dependency. Why it will affect existing projects with old deps?","accounts_in_message":[],"_revision_number":3},{"id":"7c6d4193dbce80fce9adae80e824fbb4e7e75447","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-14 20:05:49.000000000","message":"Patch Set 3: Code-Review+2\n\nbecause I\u0027m dumb","accounts_in_message":[],"_revision_number":3},{"id":"ae9555660a28f18657f0627190fdad63a9f8ddbe","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-14 20:09:59.000000000","message":"Patch Set 3:\n\nit\u0027s OK, I\u0027m calling it \u0027I don\u0027t need to review on weekends\u0027","accounts_in_message":[],"_revision_number":3},{"id":"11d5e47383cc74f4909bf971d3324b17485e4531","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-16 06:53:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a9c0e06f7f8afcc0fb542d573a126fbaf9c61652","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-03-16 07:16:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6696a39fb48eb3cd84551fbde39667f3ca5616f4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-16 09:26:44.000000000","message":"Patch Set 3:\n\n\u003e Lajos, the reason why this code is in my own repo is that I just\n \u003e wanted to unblock gates asap. I tried to summarize all issues in\n \u003e the mail[1]. I\u0027m OK to move it under the OpenStack governance in\n \u003e the future.\n \u003e \n \u003e [1] http://lists.openstack.org/pipermail/openstack-discuss/2020-March/013322.html.\n\nThanks for quick response and fix, but as I answered to the thread I am on the side to have these under opendev umbrella to have long(er) term fix for this specific issue.\nBut I can leave with it now, just keep in mind and not forget about it in the midterm","accounts_in_message":[],"_revision_number":3},{"id":"283d9966774821d181a923031c9c8e1946f7e529","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-16 12:39:07.000000000","message":"Patch Set 3:\n\nLajos, I do understand all the reasons why it’s good to have these forks under the opendev. I’m going to do it a bit later this week. \n\nSpeaking about this particular patch, we can go forward as is: packages names will be the same if projects are under OpenStack governance or they are exists only in my GitHub repo.","accounts_in_message":[],"_revision_number":3},{"id":"c45873c3c0d9294bcce2d43be8108fc16ac5680d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-16 13:07:09.000000000","message":"Patch Set 3: Code-Review+1\n\n\u003e Lajos, I do understand all the reasons why it’s good to have these\n \u003e forks under the opendev. I’m going to do it a bit later this week.\n \u003e \n \u003e Speaking about this particular patch, we can go forward as is:\n \u003e packages names will be the same if projects are under OpenStack\n \u003e governance or they are exists only in my GitHub repo.\n\nthumbs up :-) I uderstand the reasons behind this, and I would like to add all views to this patch for later references if it will be important some times","accounts_in_message":[],"_revision_number":3},{"id":"96370b3143d3ec5110a7aaaddaea3a8c496b5679","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-17 14:26:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks good. Location of the actual source can be a separate discussion and does not impact our use of the published packages.","accounts_in_message":[],"_revision_number":3},{"id":"01c68abc8b4e2d3cfcbb6a50bd90f852a7490a32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 14:26:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"85587ad0d931de047892832cd08a240bad95ac81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:30:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de0d52a71fbf45178c6dab238c0d9e14 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f392814383884c098f93a789444d41a0 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bf84040bd004bf5aba881e91d3805a0 : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44efe43ad2424714a55f0496249d51ca : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd3ede198fc34b62a2de2d3be7552062 : SUCCESS in 7m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/483ed003560149319579671462705e52 : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8617b048c6e1434b868db702afd9b4e0 : SUCCESS in 10m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/57945955064e4c998a3cb1e560aa3cbd : SUCCESS in 9m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9e57d46732e14d889c90a190e556a16f : SUCCESS in 5m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/34398d45e45245df8c426d8a6275af54 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bacedd32b0a24889b2d4835fe9822f65 : SUCCESS in 8m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4eeb5e84d56d4ff4a1d9552abce12332 : SUCCESS in 16m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9a0681c184a241a687b7d428b7689acc : SUCCESS in 25m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f647809217884708b0221911cbd6d3f4 : SUCCESS in 14m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8886c28aad234b14bbf10c36fd01070c : SUCCESS in 17m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a4d710cbf91f404297704fe7be639769 : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/40e3f67e488c4f7097d4e7bf92ca4f68 : SUCCESS in 5m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/23a9b339750e4b9c90eba982074d447c : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b01a0dac0664fe7b7cd0cdbe23d6d14 : SUCCESS in 9m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a701a973d57d441891d97069c34853d6 : SUCCESS in 12m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5cebf1cc401f4cc9b4c75c133cf45c29 : SUCCESS in 18m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8801dd90eb054a3694c018a34e96c296 : SUCCESS in 1h 05m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f24f1d225044481fa0320bd9ec19d3e3 : SUCCESS in 1h 15m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c507873e437d423b9b920f988b21cabe : SUCCESS in 1h 30m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fbaf121c2bf348c9a635f8f309f87d04 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":3},{"id":"2164a1199ed61b11ccc2fd0c2137a962c1679ba1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:30:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f613f94576fb91791711d15167b293b0bbb31680","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:35:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4fe017663b04304a93d4293ac0128fb : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"67c47fb5939fa9e84ab266253cd9868308b679c9","revisions":{"0328d413133209b165e1d9ee2d671827a967099c":{"kind":"REWORK","_number":1,"created":"2020-03-13 15:35:17.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/76/712976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/712976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/712976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/712976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/712976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/712976/1"}}},"commit":{"parents":[{"commit":"df7163634ae9916a6a6003b649ee183dc1331f03","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df7163634ae9916a6a6003b649ee183dc1331f03"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 15:34:51.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 15:34:51.000000000","tz":120},"subject":"Use pyScss2 and django-pyscss2","message":"Use pyScss2 and django-pyscss2\n\npyScss and django-pyscss packages are not maintained for a long time. We\ncan\u0027t use the pyScss with teh setuptools\u003e\u003d46.0.0, so I forked these\nproject and switch horizon to use them.\n\nChange-Id: I78b3c7f8fc6bb0759b8a4e382ad18f43622f5b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0328d413133209b165e1d9ee2d671827a967099c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0328d413133209b165e1d9ee2d671827a967099c"}]},"branch":"refs/heads/master"},"38dc71b3369b087fcd0ac2d797d519c839dd54f1":{"kind":"REWORK","_number":2,"created":"2020-03-13 16:06:51.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/76/712976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/712976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/712976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/712976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/712976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/712976/2"}}},"commit":{"parents":[{"commit":"df7163634ae9916a6a6003b649ee183dc1331f03","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df7163634ae9916a6a6003b649ee183dc1331f03"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 15:34:51.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 16:06:39.000000000","tz":120},"subject":"Use pyScss2 and django-pyscss2","message":"Use pyScss2 and django-pyscss2\n\npyScss and django-pyscss packages are not maintained for a long time. We\ncan\u0027t use the pyScss with teh setuptools\u003e\u003d46.0.0, so I forked these\nproject and switch horizon to use them.\n\nChange-Id: I78b3c7f8fc6bb0759b8a4e382ad18f43622f5b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38dc71b3369b087fcd0ac2d797d519c839dd54f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38dc71b3369b087fcd0ac2d797d519c839dd54f1"}]},"branch":"refs/heads/master"},"67c47fb5939fa9e84ab266253cd9868308b679c9":{"kind":"REWORK","_number":3,"created":"2020-03-13 19:07:05.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/76/712976/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/712976/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/712976/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/712976/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/712976/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/712976/3"}}},"commit":{"parents":[{"commit":"df7163634ae9916a6a6003b649ee183dc1331f03","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df7163634ae9916a6a6003b649ee183dc1331f03"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 15:34:51.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 19:06:09.000000000","tz":120},"subject":"Use pyScss2 and django-pyscss2","message":"Use pyScss2 and django-pyscss2\n\npyScss and django-pyscss packages are not maintained for a long time. We\ncan\u0027t use the pyScss with teh setuptools\u003e\u003d46.0.0, so I forked these\nproject and switch horizon to use them.\n\nChange-Id: I78b3c7f8fc6bb0759b8a4e382ad18f43622f5b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67c47fb5939fa9e84ab266253cd9868308b679c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67c47fb5939fa9e84ab266253cd9868308b679c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
