)]}'
{"id":"openstack%2Frequirements~713818","triplet_id":"openstack%2Frequirements~stable%2Fstein~Icc66384e4b15a6495569c0586274ca6051b2552e","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Icc66384e4b15a6495569c0586274ca6051b2552e","subject":"update constraint for keystoneauth1 to new release 3.13.2","status":"MERGED","created":"2020-03-19 10:20:59.000000000","updated":"2020-03-19 19:34:26.000000000","submitted":"2020-03-19 19:34:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713818-1584646466368-603078c3","meta_rev_id":"daa9aa276b4d266be8f587fd60b4268ab01d7078","_number":713818,"virtual_id_number":713818,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 19:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-19 12:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-19 15:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-19 15:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 12:16:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-19 15:43:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-19 19:34:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"331c4c2f502d004c79d89d2edaebd37edb5c3904","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-19 10:20:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"321e030a5bff6f15cb39d37083a753584f272e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:56:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/be3ff60c53a84c2db5657db5c9dfaccd : FAILURE in 6m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b95b008967f14bf9994d25668dadec17 : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40e4dbc8235f4ed193244952d46195e1 : SUCCESS in 5m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/44b90306c5f142b6adae5f3433af9c58 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01a73c9cf233442eb56bd9ab644062af : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c8bd617326b49d89eb138762f3d6dc8 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ba183a934a7439ba47d907e2a2388b9 : SUCCESS in 6m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/67f0be6daff14e1db739e5b22108b3b5 : SUCCESS in 7m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f0c536901ff84806bfc19bb8e291d413 : SUCCESS in 18m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/54b4c4f7a4e9428e849f4ef8c8760b6f : SUCCESS in 11m 09s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3c04e0323a564e828f46eccaaa50fcfc : SUCCESS in 13m 12s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/01bc87c79154417ba17c51523df8792b : SUCCESS in 12m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/706344ef4c444a059f878dd696382ce1 : SUCCESS in 7m 11s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/1aa725ea5b6f4585a29f2b75f4d2f6e4 : SUCCESS in 6m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0529a2eaa9ca42cf8070fb0852ccacfa : SUCCESS in 11m 30s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/eee0555d386f4198acfeff09f59ea87d : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bdd55a1af60841e6a00887ba99d2a991 : SUCCESS in 8m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/04ba3ff7625b4a82a1bee099a28e36e1 : SUCCESS in 27m 52s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/48d28a24c9884c5b8fcb910b583398c4 : SUCCESS in 29m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7c774003409e49cab1217445ce659d79 : SUCCESS in 27m 27s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/0437e450341742b38dc1e44c0ef579d0 : SUCCESS in 26m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/553129411ff3446c8cc2ee3d1f5abbd9 : SUCCESS in 18m 59s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/daa7a420ce9d450bbffa3f832cce06ec : SUCCESS in 16m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b2214e5ca334553a4f9b2c45d2965ad : SUCCESS in 22m 39s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/49157778e666470ca652176f62003450 : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a36d64215c7d4f63a2ca2ad8e0f08632 : SUCCESS in 10m 35s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b7b01971c525404aa37a190509390cc2 : SUCCESS in 9m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c02bbb771c2c4430a890f07e9599d13c : SUCCESS in 1h 21m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d365713b89bc4e2cb79f82a41f4a049c : SUCCESS in 1h 28m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6a2de662a10a417390925a4322121e93 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c00eb6171f206e90a66ad579bf5c1c405462d750","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-19 12:16:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b6e34ff90437f90d010adcd1c07dd5124d1c96a9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-19 15:43:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"76c70dec37e1d61df54dfd38c35900ef359a47e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:44:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8de9551ab5dce71c8961730c1603d0561e939366","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 19:34:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5db1c870f1094d7000e19bb80c2059ebd20a988e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 19:34:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be5e1ccf8a5542e49d64ff7037af3452 : SUCCESS in 8m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30eac7cd2d1d4d10b3f9c80d72d5e308 : SUCCESS in 7m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ac254c33137c4b5c81b93d25a83f81a8 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/172bbe41f79b4dadbf0feb1fbf18f2c4 : SUCCESS in 9m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1ab70e05c9a40eca1fc1247df52db7e : SUCCESS in 8m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d38136fb3eab42e78bfb74410abb0b25 : SUCCESS in 6m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c89352ccfdf44fe0a7e0a1e24d360a59 : SUCCESS in 9m 19s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e59c76d320f445468f563520e9684334 : SUCCESS in 10m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0ca365dc922741e7802ecdfed4cb1b99 : SUCCESS in 10m 34s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/685cc4156d724c46b21e065640fe5275 : SUCCESS in 14m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5560e37a4c264aa39600ecc0563e44a0 : SUCCESS in 7m 15s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/da84f15851094eacabafa0a4fa15c072 : SUCCESS in 7m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eec3fd11712d4dcf8360f830f5655a2b : SUCCESS in 11m 46s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/bd714a4cd1844c038d51ecab9e693b54 : SUCCESS in 13m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c285db5fdf74860b2fb9c0767320e14 : SUCCESS in 10m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/683b5486a7914ecf983418d2e8592b0f : SUCCESS in 24m 58s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/cfc7591f2d734222b9b1db782b145a2c : SUCCESS in 31m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5db5e053dc044d65a17e84407e9d099c : SUCCESS in 27m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4c65c3fbeee04426a8d535a346953288 : SUCCESS in 23m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/50329fbd113f4f099096d0baaf0ac225 : SUCCESS in 16m 50s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/4ffab37041704fa29415cebab4307964 : SUCCESS in 18m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd74d3d6a2334abc8ccfdd6c8a8f3085 : SUCCESS in 28m 32s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/44396f2558f7469fb88fabad1960b859 : SUCCESS in 9m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/83ac8037df434cf98d222f37c494897c : SUCCESS in 9m 52s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a0482345ea3944b5acfabe14ae2484fc : SUCCESS in 9m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdefbd766a78464a8b9695e196bed109 : SUCCESS in 1h 14m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/94b268ac71e24bdb87bb614ed9a4cd70 : SUCCESS in 1h 30m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e947258827c3450f8c9813f2a0361a38 : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"317b2e3d120b6448fd27ab9a2839854bc3130537","revisions":{"317b2e3d120b6448fd27ab9a2839854bc3130537":{"kind":"REWORK","_number":1,"created":"2020-03-19 10:20:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/713818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/713818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/713818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/713818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/713818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/713818/1"}}},"commit":{"parents":[{"commit":"c5206f25e2ec3a039ca8848683e027b4a8717f86","subject":"Merge \"update constraint for oslo.utils to new release 3.40.6\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5206f25e2ec3a039ca8848683e027b4a8717f86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-19 10:20:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-19 10:20:56.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 3.13.2","message":"update constraint for keystoneauth1 to new release 3.13.2\n\nChange-Id: Icc66384e4b15a6495569c0586274ca6051b2552e\nmeta:version: 3.13.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Dirk Mueller \u003cdirk@dmllr.de\u003e\nmeta:release:Commit: Dirk Mueller \u003cdirk@dmllr.de\u003e\nmeta:release:Change-Id: I3dc22cbbc55c69da2b559fb266f251e0f0074038\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/317b2e3d120b6448fd27ab9a2839854bc3130537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/317b2e3d120b6448fd27ab9a2839854bc3130537"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
