)]}'
{"id":"openstack%2Frequirements~713820","triplet_id":"openstack%2Frequirements~master~I32003146da642825d2cd59c1ad51ea24cda73e61","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I32003146da642825d2cd59c1ad51ea24cda73e61","subject":"update constraint for taskflow to new release 4.1.0","status":"MERGED","created":"2020-03-19 10:22:48.000000000","updated":"2020-03-19 22:12:48.000000000","submitted":"2020-03-19 22:10:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713820-1584655853983-b7a9cff9","meta_rev_id":"1bdac9078a5e3b8356e31950f796f70748bab59d","_number":713820,"virtual_id_number":713820,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 22:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-19 12:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-19 15:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-19 15:45:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 12:43:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-19 15:45:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-19 22:10:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a7cf22fafd2a6a1cac88b61caf26ead6562e620","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-19 10:22:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecccd177f94abe525955060d2a71c4ec0d419fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:57:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15907ec31aa4daa82ba9da8cd9b99b4 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a73bc506ce47442fa7df17764133ceaf : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8654aa0d4514fd0b9644a6eb29889d1 : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2364fc317bb4571b30c23651157bb8b : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd09375e5a1b41d7af76842310ff33b5 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5267ac3c43314d7cb890b40ebf522616 : SUCCESS in 6m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/374420fc7305459197081a2472c61261 : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/83a0fdefd44f4eb681f5f677701bfcaf : SUCCESS in 9m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c00374ccd16b48de8c29984f8b1bd917 : FAILURE in 5m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0a9c89d7ae9c4191bdfd7812d41af401 : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/250d1c3927c4467fac03333e372f6112 : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82df5a6b523f46d1a6f5f3995759c974 : SUCCESS in 7m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a2a736e8226244048ef66fbaf98161f0 : SUCCESS in 26m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/41fb22afb80442eb9e64b2bf3be7c97b : SUCCESS in 23m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/51456ac9da9e4b9488ede28252253573 : SUCCESS in 14m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ae167735c7624e4bab2b107ba37192f2 : SUCCESS in 14m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/384d522a9f794219a331e0648d5bfffe : SUCCESS in 17m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0783005410ef46028101c1d0a9ccb5fb : SUCCESS in 4m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1b5cba99c92d4d3b8ff72e795d9f0dcc : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ccfe525a6c14b4095066c3a9b4adb3a : SUCCESS in 6m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/33fc1b466e1542189a917b005775b8a8 : SUCCESS in 11m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fbcc61c7f7d344f0b2be4d2b93dba877 : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/751a1c987f4c4dd58cb8033fa7c5a168 : SUCCESS in 54m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/12b0780ec4f64d70b7a0eb930c2114a4 : SUCCESS in 1h 07m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41c654f775544bae89597e4209133a34 : SUCCESS in 1h 15m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f73cae68f4804f79b73bfbd7e73f6d73 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"921011c52a8a80fe0d37dec9375f611b66e23b9e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-19 12:43:04.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nStale mirror?","accounts_in_message":[],"_revision_number":1},{"id":"a139374a7b2ba46b321299987ee34fd8cf0e083e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:04:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48bdc87fc15d4997a90214af2f5c1eef : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8adf50146af4af1b1c1e988c737d733 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4fdeaff7be3490c87b039f24baa6497 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb3327bc1fc04a859289db8970510dcd : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/004c865518f24cc6a41fc23e32c4017a : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76d956623b8741aa8bfb5b049f87a79d : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/918b95f66219456c8f6903db523c899c : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4b428fd32f746ef8832f65f9df038d6 : SUCCESS in 9m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aa006c50230649ad996e129e572e2c38 : SUCCESS in 10m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3212f1552a3f4dc491385aa996ae88d5 : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/111a4eb4fc3b4004945729b009a36ed1 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2b0f8c453814fd2b2f280980a19ac91 : SUCCESS in 11m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cca3cd23da6e44268f2e7081600dce0a : FAILURE in 24m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/89618b1d422149bcb62601aa5e632787 : SUCCESS in 33m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f2546a1c3d354fb4a0c7407fdc181974 : SUCCESS in 15m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/462111bc941e4607a1086c83563df124 : SUCCESS in 16m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/467ef5bf68184a8a8030c1642559dc6e : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bd9831d78e4049c09296e8dc65e31c1b : SUCCESS in 5m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/34afc9ecbbfb4c038c400850004a7184 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7726cd341dbe48f6925672d0bdb252ad : SUCCESS in 8m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9fa6abbbb3e643a1bde4ca9879b9c794 : SUCCESS in 11m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8daa86193e3a4c9187772b96c27635e9 : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38a01864b56942efbf8c3e6bead3034f : SUCCESS in 1h 10m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7055af587f548018902b88ed2aa1aab : SUCCESS in 1h 18m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e28f46353224e4197a19f6cc46326ce : SUCCESS in 1h 41m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fccddbac3fb94d549a75c3ef2d01122c : SUCCESS in 56m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d4a8f46747b6955cb399d90899a7928a13466cdd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-19 15:45:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a6e32c0f67c3a7a81c5c2ac94febb58bf6a38cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:30:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8652a8e61a24ca99c1e5e05e46737ee : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d68eefc04714c799b2c02d7e6b02001 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33a97d0e2ba64e4e8f8991f6f4337af2 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a24036d03b9047ad8c75ead18d974b4d : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20981998e0ad488592f5f0b7fe2e3781 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9ad9fe393b045eca8cee21dccf5adb8 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e41542d726e64f9a9e70710492681673 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59fd9f32f7c5453c8200e088c9027579 : SUCCESS in 11m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fe000164e441486eb7ceaaefc930d919 : SUCCESS in 13m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3fa2342b8d414d278615d2b0a48df99c : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b2456f2a0d1442e4af16348694cb27bc : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e401df0083ec4d86ba3107d0deb01886 : SUCCESS in 8m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f9f7c9fea8bc4df5859d1ef16e33f9cb : SUCCESS in 19m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2a541a5d26134193b57353abdfbe473f : SUCCESS in 30m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b7766c3caa4b4e6da12210ae4c3215dc : SUCCESS in 16m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0e3d1da94be843799275b3db2d6c7b2f : SUCCESS in 24m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f9125666e2d4583a6f3e2083700b09a : SUCCESS in 19m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/56c8940a123a43dabaaa078159f90e6f : SUCCESS in 5m 46s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c330c8dab07743bd9f197959f1e45c0e : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c5df6261e914a25813ba91dadbbcb9d : SUCCESS in 8m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/35a74a5d0a7f498fb9fd8bbf590da0fd : SUCCESS in 10m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/badcc63cb2be417fb490ed7d31966e43 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fb1fa2091b644b49196db95c620866c : SUCCESS in 51m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8af28651a3fd4331ae5fc4915e2c6d5b : SUCCESS in 1h 12m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef2d0ef1e0eb4c688beed9f41725ae5f : SUCCESS in 1h 44m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3c956ced4cf24a389300e8702eb329aa : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":1},{"id":"51231e025f6e02c391130db8bd24231a51756a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 19:35:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8c799c97d5513fc85e1d1517a64d0766d47be4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 22:10:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"22ded60dd026eebcf3b5a290844282a1298c1e28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 22:10:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0590d03186948c090740125548371bc : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e742c232eb4bbc991a5d2a94bf1a96 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba3cfb8d6da94785bd88b2b40c9fe64c : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df381e930d17446e9b730469c9ed51f3 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a01972958d184d2897c672d48cb3cb30 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2bd2b0ed5094aa391b5b435b27ac4aa : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fe5b510585fd4a96b430c952f1cb6be9 : SUCCESS in 7m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/339460fa3d114290927c07a9539c8839 : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7d7d529ffffa42b98d2a0f1b4a8634b7 : SUCCESS in 7m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66e9f90ae59848a1815429fe5a3cc671 : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3295f95467c346369e5232f6b6d9d48d : SUCCESS in 11m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7bc7095f867544fe83d85939b5d1c81a : SUCCESS in 19m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/257b56217eb944c79b47378be229c6fb : SUCCESS in 32m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/588acf0449424a3684560953c1849cfb : SUCCESS in 23m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/59342257087045a9a1b70c4691d0dd79 : SUCCESS in 13m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/814aa78088b0435392d3740bdfe160d1 : SUCCESS in 17m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c7edf00dbdbd4537bc95b9895e2121d8 : SUCCESS in 4m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8e54e3b14e044f51bd983a7f37b8b861 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/936f33ec346746a59474bf04f1049cb5 : SUCCESS in 7m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5fb67f6291294bd0b504b6225f404173 : SUCCESS in 15m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6e5ee3664eec470ab9716985b80c6340 : SUCCESS in 14m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/963cd4242c4e4581980ac4b4fa459e33 : SUCCESS in 57m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c23e4e05c2e543a0991615b39aa051ca : SUCCESS in 1h 16m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0cff284a409404188b351090c7931f0 : SUCCESS in 1h 21m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/30d1335bcb03498a84ec2f71f279a42b : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9136ef6b7db3d82c84f4e6e0733ca6e37a6a9f49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 22:12:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/834263a9203f458083e451b820ba0147 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1497b32e080ed248378ece9148ce4998db6d2d10","revisions":{"1497b32e080ed248378ece9148ce4998db6d2d10":{"kind":"REWORK","_number":1,"created":"2020-03-19 10:22:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/713820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/713820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/713820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/713820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/713820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/713820/1"}}},"commit":{"parents":[{"commit":"11fa4bf17e6da8db6b348108c14ed9799b658251","subject":"Merge \"Add xstatic projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11fa4bf17e6da8db6b348108c14ed9799b658251"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-19 10:22:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-19 10:22:46.000000000","tz":0},"subject":"update constraint for taskflow to new release 4.1.0","message":"update constraint for taskflow to new release 4.1.0\n\nChange-Id: I32003146da642825d2cd59c1ad51ea24cda73e61\nmeta:version: 4.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Id2792199eb5e1843b2f2fb02b5574992e3ba17fa\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1497b32e080ed248378ece9148ce4998db6d2d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1497b32e080ed248378ece9148ce4998db6d2d10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
