)]}'
{"id":"openstack%2Frequirements~713822","triplet_id":"openstack%2Frequirements~master~I7db5a08898bc5c7248f4e8409fd0bc917049e691","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7db5a08898bc5c7248f4e8409fd0bc917049e691","subject":"update constraint for tooz to new release 2.2.0","status":"MERGED","created":"2020-03-19 10:23:43.000000000","updated":"2020-03-19 19:38:16.000000000","submitted":"2020-03-19 19:36:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713822-1584646608556-fcd7e5da","meta_rev_id":"1937768e8e7794533ab2d7e210646449161a3048","_number":713822,"virtual_id_number":713822,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 19:36:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-19 12:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-19 15:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-19 15:43:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 12:44:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-19 15:43:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-19 19:36:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77cedd924453e09a8d66af38864d96c41ddf81c4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-19 10:23:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eda8db1c18cbad2ac6dd5c4138ecd4062fcc3c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:51:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c904db6b34946d9af472e18a100151c : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e19079c3b802478e9051f49b8f685447 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c027de8dc25f451f8d96a73c40036acb : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1fb8091451b45eb89c7d8de1b42d298 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/102a5b745af64ac58388c01424d84c83 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/766e3188860749dc820e418a27563c07 : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eab787cb4e904899853a31a5df7c9b7a : SUCCESS in 5m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f7e0df6d3e14fc9adec1ce36ba5292d : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f017b93af654f629a50d820d6f2ca20 : SUCCESS in 10m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/41c2c3f261c5486b8d7ccea8fc94f602 : SUCCESS in 7m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/088dfb8c1fb6458796c315e9b7b776d0 : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7899467541ea4bcdaba789ac5225b818 : SUCCESS in 8m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/abb684438e5e4eb9aec241a4fb0495c6 : SUCCESS in 13m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6f0cd1b8870b4126b816780342a7d7b3 : SUCCESS in 28m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/83b63f0b208447a981d14e1ef80db530 : SUCCESS in 13m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b6e70dc10fee41ee92ca830bfc783106 : SUCCESS in 15m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03e0a8de4e7e47c088824a69ba78a5e1 : SUCCESS in 17m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1cc1bdc5abe94562b54af67154bc9c16 : SUCCESS in 6m 05s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f1d66a08056a4b82b2bfa916dc9b46e5 : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d072ce674fa743c781e8cb8b432bf66e : SUCCESS in 8m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/88de4ad0d6a94bf6b872306e1bfb7f44 : SUCCESS in 14m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/368d17ecefa84c888b168183ad20e321 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5cef95124634ec0b8ec460be1c00e23 : SUCCESS in 56m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/018bc917ceb7453ab80e1c3ef2c9f6e1 : SUCCESS in 1h 02m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d26e9f95ff4b4bf79d270f54dd798645 : SUCCESS in 1h 20m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1985d5e93fde49bf97859d9db3781850 : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0423ecd808d4f83181c17d433bf034a675d19ecc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-19 12:44:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcb005455f91f5617c008db6c626a6b55e0c7d14","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-19 15:43:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de322a0416cce86ca01ba61d97b04a679ccc8b2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:47:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff7453b95b3e24a6ba4768ab217d76208cc29de6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 19:36:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b4edc8731b957ae77ff8e672e16996fb4c03593","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 19:36:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f449b2fd72be46bf89ce247b92f70f1b : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb1c4975883a43d1a1feb4d553099b88 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a25601890394a2385ee49b9a5e66a74 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b022fc286cc64e458db49ccf67dee9d2 : SUCCESS in 10m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb50b036521c4712843385dc43c8d14b : SUCCESS in 5m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/64a8ce05ce304622b655fdb4d5bd7e4d : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4653fe3069624db4a043bc3b8fc43b98 : SUCCESS in 11m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eeb70fc8177c457d89cb77079ba04732 : SUCCESS in 12m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/186a829b8a6b4dca85380dedc5abd8e8 : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1a19c182540549c2ab0dd2ff6e6484e7 : SUCCESS in 14m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99d769d58c4540fb89bf2727635577a0 : SUCCESS in 9m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/71041b6afa21432f89fa6f33e83795db : SUCCESS in 17m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/61cb0ad823414e6a935fc28d596ad489 : SUCCESS in 28m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e5faddd0768649d28c930813ff0c1f66 : SUCCESS in 14m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2c77671b06a14231a2a0de52ff937c6b : SUCCESS in 15m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/822a20fb96114dee960f126ba7f1dc1c : SUCCESS in 22m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0f324ef98de140f08e606853a9770b39 : SUCCESS in 6m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fbda02ed37d44fbdb66e9d189ef5bde3 : SUCCESS in 7m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17c6574edcb94ec4a84352c93ea7d1db : SUCCESS in 9m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/213d2d13a6cd446c9df3c90c0271a927 : SUCCESS in 12m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6a28d52afc7f4d3386857f5f09e5af0f : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/456b8b2704124314a76d6ad11fe07129 : SUCCESS in 1h 13m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79f5822eb3254c67a2f5360d7dc74c1d : SUCCESS in 1h 07m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5ed032e53e9491e9b91ee920b1d2770 : SUCCESS in 1h 49m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e4b0e6e7d677488f92b3000432ec4ba4 : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2d72fe797ac6e8f028caa1d3c0908019fc1589dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 19:38:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2148d75c3b0f42bd9bfa6038c1739154 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3b4f8e840f9a5a6ca1c55d8f1c1fa9a0069b88e","revisions":{"f3b4f8e840f9a5a6ca1c55d8f1c1fa9a0069b88e":{"kind":"REWORK","_number":1,"created":"2020-03-19 10:23:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/713822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/713822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/713822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/713822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/713822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/713822/1"}}},"commit":{"parents":[{"commit":"11fa4bf17e6da8db6b348108c14ed9799b658251","subject":"Merge \"Add xstatic projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11fa4bf17e6da8db6b348108c14ed9799b658251"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-19 10:23:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-19 10:23:41.000000000","tz":0},"subject":"update constraint for tooz to new release 2.2.0","message":"update constraint for tooz to new release 2.2.0\n\nChange-Id: I7db5a08898bc5c7248f4e8409fd0bc917049e691\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Id2792199eb5e1843b2f2fb02b5574992e3ba17fa\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3b4f8e840f9a5a6ca1c55d8f1c1fa9a0069b88e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3b4f8e840f9a5a6ca1c55d8f1c1fa9a0069b88e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
