)]}'
{"id":"openstack%2Frequirements~713825","triplet_id":"openstack%2Frequirements~master~I353a1061be82b26ec80aad356c058d7459d9e156","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I353a1061be82b26ec80aad356c058d7459d9e156","subject":"update constraint for oslo.versionedobjects to new release 2.0.1","status":"MERGED","created":"2020-03-19 10:27:52.000000000","updated":"2020-03-19 20:16:40.000000000","submitted":"2020-03-19 20:14:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713825-1584648883799-4c3f8529","meta_rev_id":"e8d718e9cb22905e4225b77280e7382c8b299a2b","_number":713825,"virtual_id_number":713825,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 20:14:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-19 12:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-19 15:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-19 15:43:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 12:44:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-19 15:43:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-19 20:14:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59de328855d1ddb4c851556a33bdbb2644a79e75","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-19 10:27:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e2de4137fb5f1a7515c0601c6365d80b87723a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:11:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f6b7f5018df4c0d9916528e254e7590 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e3d0cb54b4447f994d8f5e51550596 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6660cc918cd845eba51241e42884a0e0 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d09dccc9b81349f8ac3deb692d1990c1 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c220a11c284170bad29b2e940b23f6 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c24a2ce0b774a8291e6126094d55490 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25cd69911e5d400a9560dc8b30e4fd51 : SUCCESS in 5m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c2a4c9e392514174bda83208762c5915 : SUCCESS in 11m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d4b042983f204952a13fe0e9df97e813 : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e62560b5b528443e87fc1c9cb6812ce3 : SUCCESS in 4m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/482d2960243a4ec38f8e64861d37fcd6 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b17f400186f14d8a874f2b2a09d19e59 : SUCCESS in 9m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e4b6526a720d4e7498aeb1d002e27151 : SUCCESS in 12m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0f1d8f9b608f464b89e9da04a2ba7ca5 : SUCCESS in 25m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d1df6a2726a14942802ea3aa56cbfa75 : SUCCESS in 12m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/12d727ba100345f992e00f08cd97ef31 : SUCCESS in 14m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29cbbc77d7484ab8912771121d1c18d9 : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/09e9876375f645d9a47015839b2ecd84 : SUCCESS in 5m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6e513fc015284cbaa3008e89eb4dad5f : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90c083abdcdf47c49783d5f7cfc91727 : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6379188988754eb6913d120e914bde27 : SUCCESS in 11m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5eb69148112747c6aa303ee2f09034f3 : SUCCESS in 16m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2806aa090c441c8af885aa5fce7a95b : SUCCESS in 53m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b56e0e09483d4045be96c9f02aa65f81 : SUCCESS in 1h 05m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1bb9ae6e80124fac95aa5a5e0c77cb2d : SUCCESS in 1h 36m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9f34a63696704ee4a5ce8d5daee48c4e : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0632513a935485a34466b76d838d46c8810a4d06","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-19 12:44:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8741c6195c31e1e3c5b69263ccff7e7b538a6e96","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-19 15:43:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa7ea6b9f413eb080f9045d16416e5c29723c1ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:48:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b5c6e6a1c32b55fdba5f98b0e2fbdd050f428fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:14:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3abb648cb709fe740b4b556e612b37b118987143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:14:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7a3bdfeba104dbcbadeed1a9f9b9e03 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94b089ced6848a6a4108a72b6dbf1a5 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8891ee45dea4aa0932dac1d5785ab29 : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f24381ad4de49caa53324145792e2d1 : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98af2a3302784f9d80f4272617c4ff80 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71ebb54fafbb40918ad6817b67e7db12 : SUCCESS in 6m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/74edca5f20094106ba7175cbb3dafc90 : SUCCESS in 9m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4827eb07cfd1400d9ef0b234312f632b : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3ca9c4a0b78f40eb90c66208d59d2640 : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3c247fd974fb46b2bfa9c81212f97066 : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/739c866c41084c20b32fa84c3838fa0a : SUCCESS in 8m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/171907ef52804df6b156245dbb5382bd : SUCCESS in 14m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/986509c4822447409f42e50cc42cb7fb : SUCCESS in 23m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cee8d44570574e6e8ff859d5b38a179c : SUCCESS in 24m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4c1e57bb11194cde95fdd4c184d0963e : SUCCESS in 14m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b307bca2f8704e31904715520898ab59 : SUCCESS in 23m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/16d782c9c952487a8a544eb931144ce8 : SUCCESS in 5m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d1ad36f1f1a74c65a19b0f8ee99cafa3 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7179448640384c099dd02c30db069b3e : SUCCESS in 7m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a415fa77d55b42e4af886f7efc78212f : SUCCESS in 10m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c3000747a11a4e1a9683ce945cb5989d : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11054b3a84e7437aa1d8f166ebe93c85 : SUCCESS in 1h 04m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6985f76355c8424ea071c666bd7fe17f : SUCCESS in 1h 09m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7b51c579f2e42cfa446d0d1736288b5 : SUCCESS in 1h 18m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/10f05e01afa94411b229104964185bff : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8a17cff50793c6790d1f4134e1903c142dc0934a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:16:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a516b25f4414623b4fd41378ec20cf2 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34721b8fa88e233f7c8d0106f4d201105a4662cc","revisions":{"34721b8fa88e233f7c8d0106f4d201105a4662cc":{"kind":"REWORK","_number":1,"created":"2020-03-19 10:27:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/25/713825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/713825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/713825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/713825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/713825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/713825/1"}}},"commit":{"parents":[{"commit":"11fa4bf17e6da8db6b348108c14ed9799b658251","subject":"Merge \"Add xstatic projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11fa4bf17e6da8db6b348108c14ed9799b658251"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-19 10:27:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-19 10:27:51.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 2.0.1","message":"update constraint for oslo.versionedobjects to new release 2.0.1\n\nChange-Id: I353a1061be82b26ec80aad356c058d7459d9e156\nmeta:version: 2.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Id2792199eb5e1843b2f2fb02b5574992e3ba17fa\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34721b8fa88e233f7c8d0106f4d201105a4662cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34721b8fa88e233f7c8d0106f4d201105a4662cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
