)]}'
{"id":"openstack%2Frequirements~714035","triplet_id":"openstack%2Frequirements~master~I663b3562b2668e99a84063babd6b9006acfeff34","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I663b3562b2668e99a84063babd6b9006acfeff34","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-03-20 06:24:06.000000000","updated":"2020-03-20 10:08:45.000000000","submitted":"2020-03-20 10:07:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714035-1584698822464-baf8bc82","meta_rev_id":"89f4d322d0165400dc7cf72fefbe586b47c3c02c","_number":714035,"virtual_id_number":714035,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 10:07:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-20 07:17:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-20 07:17:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 07:17:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-20 10:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34c167c3dec4c66fb6cb127dc10f8aa17f90cfbc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-20 06:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b1d2389608b2727e38973efd78947289bfd0021","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-20 07:16:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7bc4173a33f925ac3a2fc94efdd7ef3fd1e550aa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-20 07:16:46.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"2c2e76addc002b97e9a082cc7ed0c8956da1c81e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-20 07:17:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0014f01fb1d841b72b2ee1a12f47a334c3b8e233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 08:47:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe107e3b12d4595a2fd69c132d7e449 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/273b5e83d73244cab45e8f99ed3b04d4 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/988de903d93f45d0ba3821a14cfe4fc5 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b06067dd69e42219344348a3dc8e997 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c05768d43844256b519ba724e9491a1 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57f6aeb306614933a4c9a8db393a5917 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec4a648fa9bc4f23928c1117bba406dd : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1cda9afb424b469895df687e4de17dbb : SUCCESS in 7m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2430ce2df0e44565a3ed945ef024d189 : SUCCESS in 10m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bc694e9dc7d64e4ab1dba1a6ba41ca60 : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f0518bac90064755aab7d6e5b3d9ee8c : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4928a21825274618b3a201ca7e153ae9 : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3e497af838c342ab86639dcf6e9f7966 : SUCCESS in 13m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d12e9423e7d84d31932a296596dddd11 : SUCCESS in 27m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8cd472801da345c6902958caf3a34ae4 : SUCCESS in 13m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7f2778d146224159a707840c08d5c6d8 : SUCCESS in 16m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de1dc0f9910c47fabd83f6f64b7f6ec6 : SUCCESS in 16m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c018778934e2445c8c91e6d49ea5aa39 : SUCCESS in 4m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6f0b9d97476244daab1ac8a02071d3b3 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/feb02d6988c5495ca24e70f39c5b3e73 : SUCCESS in 7m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d2c5d92eb751486f8e5e87b9754ac975 : SUCCESS in 10m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d3203b500f9d48f093e99e8e2adc2c32 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b988730d11d04b23925d07d42a6e6d51 : SUCCESS in 53m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bc6abbeb07e472ca9568261ca6531e7 : SUCCESS in 1h 13m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7511f4ff83e043f0a9e0c5e52ce55f05 : SUCCESS in 1h 25m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/868a361efe4a49299c4509ca4571e073 : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c7cd224156c6e4f4d4bfd9fddb33c6239bdc868d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 08:47:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e1c1f349c186fc3a638ceef0defc97925e3cb37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 10:07:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/568de67748464b559acf1ab1aaa579ad : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b94e95570de44378c58c1160c2424ed : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8829ba6f8e19460690d1d32b1db9560a : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3be19aaf80441b1a1b1be6800addb4f : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91414f4f41004c879e362cef5b2e10e9 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c9959a53c2b4b13ba2a36f573694947 : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15576dec9a3d4c1793134b37016b587d : SUCCESS in 7m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7445e114b6404479bdb81e0d222d5e03 : SUCCESS in 8m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0b4275ac651a4fe2924af41b1580f324 : SUCCESS in 4m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d8e9172a80cb4eb89f4c8a2b137257c4 : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0bb31a7e9b84e62915f3e5933ed09f4 : SUCCESS in 6m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cc9e1b3aa7da485197f6c9d982a3102a : SUCCESS in 12m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4bd26a521aa34f309dde17994909035d : SUCCESS in 22m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b4c8b962621140e495c48797c5453f3e : SUCCESS in 11m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4afba912c2eb476b822b10c35040ec64 : SUCCESS in 12m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44eff4bc34814676aa224628f96eff6f : SUCCESS in 17m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/909c0f51fdb2452f824329c418cfba79 : SUCCESS in 4m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/311b4b1fefc04efcb682d8df846945e1 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a6a09692fa94a3eac905d5a5bc8bce5 : SUCCESS in 6m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6b4fafed02914b8698fd9a4a3d6f1093 : SUCCESS in 9m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/44903b84f5244ef7ba5fc6335de9a9ba : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48dcc2c49afd4c149a1f6b8499b01160 : SUCCESS in 58m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80710b208e064150a4356eeaeffdaf62 : SUCCESS in 1h 13m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6810212e89f8441a99afa1a56f1230fe : SUCCESS in 1h 16m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/37108b6c6bd04d3496111f3b13a06e40 : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d957560358da508da65fe99ccf542aaea94a7e81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 10:07:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"89f4d322d0165400dc7cf72fefbe586b47c3c02c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 10:08:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037eb77b7c4f40668bda5c71acb008b6 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02df513aa8c6de4469c5d5214b1dd2591aeebe1a","revisions":{"8b291ea4505b421aa326e78ce3c463e133f05d7f":{"kind":"REWORK","_number":1,"created":"2020-03-20 06:24:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/714035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/714035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/714035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/714035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/714035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/714035/1"}}},"commit":{"parents":[{"commit":"20099e615193e236915d2cffbf1136b1e2c15661","subject":"Merge \"update constraint for tripleo-common to new release 12.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20099e615193e236915d2cffbf1136b1e2c15661"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-20 06:24:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-20 06:24:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I663b3562b2668e99a84063babd6b9006acfeff34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b291ea4505b421aa326e78ce3c463e133f05d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b291ea4505b421aa326e78ce3c463e133f05d7f"}]},"branch":"refs/heads/master"},"02df513aa8c6de4469c5d5214b1dd2591aeebe1a":{"kind":"REWORK","_number":2,"created":"2020-03-20 07:16:41.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/35/714035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/714035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/714035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/714035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/714035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/714035/2"}}},"commit":{"parents":[{"commit":"369b7f60e73479403168bcf8328a17842eaa7247","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/369b7f60e73479403168bcf8328a17842eaa7247"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-20 06:24:04.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-03-20 07:16:33.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I663b3562b2668e99a84063babd6b9006acfeff34\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
