)]}'
{"id":"openstack%2Frequirements~714080","triplet_id":"openstack%2Frequirements~stable%2Fstein~Iab8237390093cc51714c8b99d573719acf4a2905","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Iab8237390093cc51714c8b99d573719acf4a2905","subject":"update constraint for python-keystoneclient to new release 3.19.1","status":"MERGED","created":"2020-03-20 11:00:56.000000000","updated":"2020-03-21 06:11:15.000000000","submitted":"2020-03-21 06:11:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714080-1584771075410-a663f06a","meta_rev_id":"6543d53c284cd64334e853961a97fd7d35b2a43a","_number":714080,"virtual_id_number":714080,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-21 06:11:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-20 12:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-20 17:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-20 17:52:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 12:34:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-20 17:52:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-21 06:11:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07bc8a2cc25b79516e810f20cd1ece175480c731","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-20 11:00:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0a8cbd3e9acb430ca1bdc6e5e6f19b51e179e49","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-20 12:34:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5597d055062dbf6b32463f031a0af5a1949931c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 12:39:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/37e65753f18a414498308d666268f500 : FAILURE in 5m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65371d8be4d64ca0a295836183706cf7 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4e4550bd00948f2ba59c8b2072e261c : SUCCESS in 6m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ac116c1f04b24d5f81cd639f359db5c9 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf767f8d415444338646fbc81ac75f54 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2e1ca00af7c401b9b055078ae0c86ac : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ed682512d7c4112ba475b19cc3c869a : SUCCESS in 2m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/208c246f369c4b64944aafcaca8ab0b4 : SUCCESS in 7m 50s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3f7336ba0551436caab1aece7802b8bf : SUCCESS in 9m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1d9c2c9153b94f0cac3663e941f5282c : SUCCESS in 8m 36s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7cc4929d522842fc9f6f7f75dc73e6bc : SUCCESS in 10m 24s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/6e2a713aaaac4cf596246971d8301a67 : SUCCESS in 10m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b7965d539fef4b4abc46f2aed0607f58 : SUCCESS in 4m 50s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/fc6560cb2a8140c292f786f5714187ff : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9e5a7348bf834999af9186be506db1d0 : SUCCESS in 11m 09s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/b4035223908d4caea4da401bdad931bb : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c3a4b1849f7b4e039af7944353170377 : SUCCESS in 7m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/76e4c25b119c4610b7377b74541bdf0b : SUCCESS in 22m 39s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/fe4a5f73655440ae9c05ef672e912008 : SUCCESS in 26m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c585d486d8e74404929bf1e481b9a6a7 : SUCCESS in 22m 38s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/cfb0d517e41543a1ade04bfdf246b639 : SUCCESS in 25m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c0dcc806254d43c8a388ac5294e197af : SUCCESS in 11m 42s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9273c071619f434e8d37c22a74c72f5a : SUCCESS in 15m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d041d9149aee4bd2b5977b9082bc1f1e : SUCCESS in 19m 13s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/3af8547fa52b41cd97483f8fe0bbad5a : SUCCESS in 5m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6e2c98eef3034de4a6e6acc489cde1ae : SUCCESS in 10m 17s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/aa710d02711a4e2899e2e879c98cea46 : SUCCESS in 8m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f4d4af0f85940809bc69cc6eb66d6e5 : SUCCESS in 1h 07m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/806f8190f8154dd4872eb2ed5a8285ec : SUCCESS in 1h 33m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e0b121af1d1948be81f1320afcc4df59 : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4abfc382a548da56ea5b82e02c708cef05dbe1f3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-20 17:52:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"57fb2def62f4dec051d047af7defd95e9032433c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 17:58:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5df61f6b924a6ef183e14bb39014a34b983c9b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 06:11:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f1fe1e627d24e61b431eb2c0d2f7f53 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8b50c63294f4e71b99ac45ecdefadfd : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/18241f21ce964ff1ba58aece180d11c5 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4209a704c0f4bc6ab9629a65b0877c7 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d90bf79edd24b8da533bea88c32595f : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04b8afddf20945ac93c83936da09ed4e : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ebcbc47dd7564f22b3deafad9e9a2a39 : SUCCESS in 7m 32s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/cf664e419bb444d0bf5bdf0ad1bac4d8 : SUCCESS in 7m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c10e78218e054c9fac836df50c49faab : SUCCESS in 8m 25s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3b02a97398f8412eb1a9d472f1497aee : SUCCESS in 9m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4526d783b74f4025ad532c7663102195 : SUCCESS in 5m 08s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/bd5153469a724b9aa528a5d470dd54f2 : SUCCESS in 10m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c514d4ed58774989b18a3691bc6f8437 : SUCCESS in 12m 09s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/503c3c6fe85d44af9aff554f871244f3 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5587d5b946b4cf5af76a2fa507dbe9a : SUCCESS in 6m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/46c92672142144748e5d938680a8605e : SUCCESS in 20m 16s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/6b0a652d1bdc4b2b8eaf1ea4760dfeb2 : SUCCESS in 25m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/903bb2d392994c28985e1a821f1fc26f : SUCCESS in 29m 50s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/3dcaf1234e6c4377b14f426cab98b913 : SUCCESS in 26m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b37ac28244084ebdbb708bc0b7b92fc7 : SUCCESS in 11m 59s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/7379b84bdd4a40e7803b8f38505213b8 : SUCCESS in 12m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c5fc40193d54bc9be50f55941bf5003 : SUCCESS in 24m 59s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/bf844c1d3b0f440b9cfa4b6dadb25fad : SUCCESS in 5m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0611fbc6fc734de08f29e87907fd11f9 : SUCCESS in 13m 17s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/63659681d5c6451ab1b7a913987f60e1 : SUCCESS in 7m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/90828403960543e68b6bb5425941b13b : SUCCESS in 1h 08m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dbc8db2d81cf48b6a21728d108092d09 : SUCCESS in 1h 31m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/18ec9b7670e14ac59869c605c3ae2a69 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b67061c1807e133c9143728b79c3085b8ff9e16b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 06:11:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cebe7cb0f5c37bb8fb36600f098c5859670e931a","revisions":{"cebe7cb0f5c37bb8fb36600f098c5859670e931a":{"kind":"REWORK","_number":1,"created":"2020-03-20 11:00:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/714080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/714080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/714080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/714080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/714080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/714080/1"}}},"commit":{"parents":[{"commit":"317b2e3d120b6448fd27ab9a2839854bc3130537","subject":"update constraint for keystoneauth1 to new release 3.13.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/317b2e3d120b6448fd27ab9a2839854bc3130537"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-20 11:00:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-20 11:00:54.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 3.19.1","message":"update constraint for python-keystoneclient to new release 3.19.1\n\nChange-Id: Iab8237390093cc51714c8b99d573719acf4a2905\nmeta:version: 3.19.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Javier Pena \u003cjpena@redhat.com\u003e\nmeta:release:Commit: Javier Pena \u003cjpena@redhat.com\u003e\nmeta:release:Change-Id: I30eff07200e674f664b2139e0065c3aba494e527\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cebe7cb0f5c37bb8fb36600f098c5859670e931a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cebe7cb0f5c37bb8fb36600f098c5859670e931a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
