)]}'
{"id":"openstack%2Frequirements~714191","triplet_id":"openstack%2Frequirements~master~I96f28c282a4aafd7ba7bd02770cafd7d49c90b0a","project":"openstack/requirements","branch":"master","topic":"oslo.policy","hashtags":[],"change_id":"I96f28c282a4aafd7ba7bd02770cafd7d49c90b0a","subject":"Blacklist oslo.policy 3.0.1","status":"ABANDONED","created":"2020-03-20 19:02:20.000000000","updated":"2020-03-24 16:28:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d15c82cf1712cbdac22771437150bc20422a9377","_number":714191,"virtual_id_number":714191,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-24 16:23:29.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-03-24 15:31:41.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-20 20:41:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 20:41:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 16:23:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c18affc382f1ae0a4eea004ebb4f32ee13bd11ab","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-20 19:02:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36a67d1bab1e209ec8de7857e132ec529fda715a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 20:41:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f05b0cee1034715a85c00b57264fb24 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b9345fb61949eba538d6e73a4c9d09 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0b640a6c3054b64bec09efbde1add79 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bcfeb5cb0fc43a3ba825510586f523c : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12eae73a9efc4efe90c2d1eb16155953 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0b5282cbcbd4966a0ae49aa3446c19d : FAILURE in 4m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9896601061d94d1895c1446944f3ffd8 : SUCCESS in 1h 04m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81304463f91c4017ad6a195eef9a9fd6 : SUCCESS in 1h 15m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3415e5e0f134de3af56931a3ebc925b : SUCCESS in 1h 37m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d2f486262411409193f9811fe1874a62 : FAILURE in 44m 41s","accounts_in_message":[],"_revision_number":1},{"id":"334462ab60d6fc108e5432c304174b6c43c65746","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-24 15:26:14.000000000","message":"Patch Set 1:\n\nrecheck\n\nis there a upper-constraint that needs changing too?","accounts_in_message":[],"_revision_number":1},{"id":"127e77c3b80664d23dd1a84f90470b870afc6c97","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-24 15:31:07.000000000","message":"Patch Set 1:\n\n\u003e recheck\n \u003e \n \u003e is there a upper-constraint that needs changing too?\n\nI think that was addressed in https://review.opendev.org/714207?","accounts_in_message":[],"_revision_number":1},{"id":"fe2325765b4fd401160d61f4a407bc89001db787","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-24 15:31:41.000000000","message":"Patch Set 1:\n\nI\u0027m tempted to just abandon this one now that we have a better higher constraint. Opinions?","accounts_in_message":[],"_revision_number":1},{"id":"d06fb31b44c98caa87eec615bbfee8990dbb8f8d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-24 16:23:29.000000000","message":"Patch Set 1:\n\nabandoning is fine with me, especially as it\u0027s more of a \u0027dev\u0027 version (master isn\u0027t \u0027out\u0027 yet)","accounts_in_message":[],"_revision_number":1},{"id":"84ff00adcb19c0bb870b9a78d4e29e5145528272","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-24 16:28:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c988cf3fdd362ab32ba3eb9b15b68802e3c5dee","revisions":{"3c988cf3fdd362ab32ba3eb9b15b68802e3c5dee":{"kind":"REWORK","_number":1,"created":"2020-03-20 19:02:20.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/91/714191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/714191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/714191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/714191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/714191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/714191/1"}}},"commit":{"parents":[{"commit":"02df513aa8c6de4469c5d5214b1dd2591aeebe1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-20 19:01:32.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-20 19:01:32.000000000","tz":-300},"subject":"Blacklist oslo.policy 3.0.1","message":"Blacklist oslo.policy 3.0.1\n\nThis version has an issue registering CLI options that can cause\nproblems for consuming projects. A fix is on the way. Until then, let\u0027s\nblacklist this version.\n\nChange-Id: I96f28c282a4aafd7ba7bd02770cafd7d49c90b0a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c988cf3fdd362ab32ba3eb9b15b68802e3c5dee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c988cf3fdd362ab32ba3eb9b15b68802e3c5dee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
