)]}'
{"id":"openstack%2Frequirements~714207","triplet_id":"openstack%2Frequirements~master~I9922429bc3d9de60d1bd427082c5cc37c88db9e4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9922429bc3d9de60d1bd427082c5cc37c88db9e4","subject":"update constraint for oslo.policy to new release 3.0.2","status":"MERGED","created":"2020-03-20 20:03:08.000000000","updated":"2020-03-24 01:14:42.000000000","submitted":"2020-03-24 01:13:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714207-1585012405584-a663f06a","meta_rev_id":"1ec42fc90b22083afc1c2e23431f72b0fa8a30b4","_number":714207,"virtual_id_number":714207,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2020-03-23 19:30:54.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 01:13:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-20 20:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-03-23 20:55:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-23 12:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-03-23 22:21:55.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 12:52:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-23 19:30:54.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-03-23 20:55:59.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-03-23 22:21:55.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-03-24 01:13:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22fc915357eb861fd4f251aaed8020ace9a6cb7d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-20 20:03:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"647484587e9e6a5d8f5d508299b62f1a7a939ef5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-20 20:03:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"733aac4dfface20f7a90adcefbc3a79c3a6fbd71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 21:22:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c4f5cc59374c07afaf9d8f89d30964 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d1de2f755d43c3a213a1105b56c7db : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f89f1d5c2d9e479c8bbe4695be902bbd : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3218b0f6697a404d9989f005cb636cb9 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56ea120314a6472f8ba4f45bab7af23b : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2075e52c0c8c44829ff553ce29cafa19 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22974087a2a8416e93cff50418c21ad6 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f419e135cf6044698c293e5458ceb280 : SUCCESS in 9m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2d055cfe3fca4cadb7fa9aba13d838c8 : SUCCESS in 9m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e38680583095474b80231319fce3ce00 : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/80324468a89940adba58a5ed893a6cb0 : FAILURE in 6m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f615c314fff64660b92b1eb57c64b15e : SUCCESS in 8m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f5a416af42a8433585456dd32082845e : FAILURE in 5m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f88c24a1153046038412afa45e287e1b : SUCCESS in 25m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f94207789486405c9acc72f2d2e6646a : FAILURE in 5m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5eef08f1dc114224aa81876685c2f48d : FAILURE in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30027c30f6ba4f4eb6103a0b4df1c0fc : SUCCESS in 19m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9fa2b769283f46fbbdf6c7e4c6c13aa5 : SUCCESS in 5m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7c495e3fdf21455eb2c7c405b8546f67 : SUCCESS in 7m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/433d297be8c041149caabc8eeadd4d5b : SUCCESS in 6m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c129d730de494c5a8ff0cb604b18f4d0 : SUCCESS in 11m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5766f50181104d0a8e5da817995a8446 : SUCCESS in 8m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4ea2d8529454aba87b8b8cfd6c7d279 : SUCCESS in 1h 03m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a8b795d7f8b47eba1769fe22bf32933 : SUCCESS in 1h 11m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e60f44edd4d1400a877c976353284a54 : SUCCESS in 1h 18m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fa6bf3c31526453294177de2ac06cba2 : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":1},{"id":"77684188d93c3cc7e9b85d81066580aefd75e8da","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-21 20:15:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d0163e8e21ebbd865167938a3e865207c303c359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 21:38:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d545e44e07384fb799aa7702ff1407d9 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae17c2127a544039a14d83c540a4bef9 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0f5c19351714b4a92758977860b9ad8 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8d519096c044e71901266b5204c2408 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a89c10a8f77747328d43832aaaeba30e : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b894e50d4b214732a3e4c72717833430 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95dd9477cab447ed84dad6b64a8c089f : SUCCESS in 2m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/31306cd1d80b4aeb9f14f671fdd7ff23 : SUCCESS in 7m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1e7bf83ad99d44a3ad4c094389616db7 : SUCCESS in 7m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a1ed1b9ed23643c2bf2149eb8d9ca6b2 : SUCCESS in 4m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/be4d61265b0f4bbb9069b150cbd8a9f7 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff1e9da19f824e6b8bd03f03897cafe4 : SUCCESS in 6m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/03b1cb43e9bc47b6b618132bcfc189d6 : FAILURE in 14m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9c7cea1aedea4d6a8935422050cb6b4d : SUCCESS in 25m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5bc3c03fb44741bb9265cfa1f8302eeb : SUCCESS in 12m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6ae30a7820024276acbf2880dcf59193 : SUCCESS in 13m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6cd92a10b394fa685f9d0257830b706 : SUCCESS in 15m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f2a8e2fd18374db98e7d744284fcf336 : SUCCESS in 4m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8f2e56be64d54488a55145d2e2b587b0 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9bb75efd3fe7492db07ea48c4272053e : SUCCESS in 5m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5f1c151121a04bc788e5bbdcf1e2aee9 : SUCCESS in 9m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1398cdf34d8645f7a0428284503324b0 : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6541a944bacb42599d619bec6a5a1d2a : SUCCESS in 56m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a872d09f2cd049f5bc5d9a3b76522e0a : SUCCESS in 1h 04m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0090e484393b410abfa6c7c30aa381ac : SUCCESS in 1h 18m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/21571e0b1374461a809607f60be25a5a : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e1bd934b35d1b2f6de838b5c5614a007dab771c6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-23 12:52:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1e2538872b7d7cf8e2af07e1bcccb87845e9f0f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-23 12:52:53.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"541ccfe09e418d55560b53ccb944a5b340e855d4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-23 12:55:03.000000000","message":"Patch Set 1:\n\nDon\u0027t see an obvious connection between the failing keystone test keystone.tests.unit.test_v3_auth.TestAuthTOTP.test_with_passcode_in_previous_windows_default and this patch.  cross-keystone-py36 failure on PS1 was a mirroring problem, tests never ran.","accounts_in_message":[],"_revision_number":1},{"id":"1ddf4d27f599a09ca87337ef60c5facab7edd796","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-23 12:55:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a92f3d4626807b0b6b8027cd8ecc8d20faf025b8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-23 12:56:10.000000000","message":"Patch Set 1:\n\nShoot, had a race condition with Sean.  At least I only wasted 3 minutes.","accounts_in_message":[],"_revision_number":1},{"id":"15037146b624b31e010921c29e6778023aa5ba97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:10:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ad94811f13470182a48035724b50e9 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e2229e42ddf42eab0839ca9ac83a558 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4c6b2c273984b4d8f3f247d065f4397 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6aa1abd3c21482a84bdc8a1270b71be : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea51889a65f44c6ca1ca4d2107e375c6 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d34ee205e9e24a94ae5679e0d726a50e : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b49b53f0fff43c8b9a0734186632bf5 : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/112ccf60396b402cb35d251a1979de80 : SUCCESS in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/11115055e10f4069bdc54ca13836180d : SUCCESS in 8m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cb7d08a32e9647058b7b963cefea9783 : SUCCESS in 5m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d40780fade2a46c3a19d9518a94bda7d : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a848b22dad94a2990e91033799dea5d : SUCCESS in 9m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/71f19f63835b4e96b66e3510a0d633ed : SUCCESS in 19m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/617e584a8c074715ad5e90d2700dcdc7 : SUCCESS in 25m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/532afc31f9b84d099e0b3b1f08e69a3b : SUCCESS in 11m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9082e4acc61545c996251602f23996a4 : SUCCESS in 14m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9951fc69de5f48c780b4c6c30e7576b3 : SUCCESS in 17m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/02cc136e27fc4c2f97f5c1355bdea026 : SUCCESS in 4m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7bff244fea1649818b402aaf647489c7 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59cf8ce7d7bd4c68947638502f9e03b8 : SUCCESS in 11m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c3678175817c4f748643213df5923327 : SUCCESS in 10m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2ceede352bee49dfbff1c0e81f96d230 : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d1171993cfa4dffa624264673140318 : SUCCESS in 53m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5fa28363f1494fd0bc574db577e6b5e7 : SUCCESS in 1h 09m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3781f1147684488969b4175eda1fb83 : SUCCESS in 1h 16m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/06a1530d294845d9b2c408f78344cbfc : FAILURE in 1h 03m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f4b4653b6f6653d0bb428cc0e79b0bc26a1966ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-23 14:12:15.000000000","message":"Patch Set 1:\n\nrecheck\n\nAnother random failure.","accounts_in_message":[],"_revision_number":1},{"id":"97c44961ab738f3b8895dbf9fef236306db6a029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 15:43:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ee211a23f34e36a408df85ec0a4e7b : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504fae48628b4d45b4eca9a23eeb891b : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b1ca442c9a4407ba5a782560622382e : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0d13f51f7694360878f2e3f4898b069 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d779702ea3014a2b920b3cae384977d2 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ffc4aa5b4174ba9a09478ae6ae0908a : SUCCESS in 6m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c8b3d745d29454089377e9acfa98e31 : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9993aabcff2446d86aa0adfcf1ac45d : SUCCESS in 9m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4051f8e4b92f43348be8dcf2f1b07446 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a37b1663bbb0446c8eb4c4c6e9373856 : SUCCESS in 8m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/265d004fbfc444dca7132ef6662bce39 : SUCCESS in 12m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ca1f2b317f043fc990e5bbd6bbd31b8 : SUCCESS in 10m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7fc6e909096f43e68c6b3651bc0d2e13 : SUCCESS in 14m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d017595162e48438637ba4ca9daf9e4 : SUCCESS in 30m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bb0d9001428f4235abf32a8efde0a4d2 : SUCCESS in 15m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5db948d137e34a71998816fd75251164 : SUCCESS in 16m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e68dfac1b4b54c4a96292fbebf8d5be7 : SUCCESS in 17m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/09af22855cc849b696999647f0eae8e7 : SUCCESS in 7m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/798731631af841cab14eaae40e7e1081 : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/757ce53101e14592af0b33aa83ddfc26 : SUCCESS in 9m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c6c5daa2b55d4513b0002ddd3718d654 : SUCCESS in 13m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/acb21042349240bdbd5c8cffdd01cb90 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51caab5b7c6d4e73b0c68547796bdc73 : SUCCESS in 1h 00m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3659a06e1c54f8f9d84cfb47bcb7470 : SUCCESS in 1h 14m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2c902394ed841ba84b3bd76466b13d8 : SUCCESS in 1h 24m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/26fc2b5a96804138b69df84e8e189aaa : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":1},{"id":"18fe36dd650242a660da60960486a543691e05d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 15:43:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f93b4231908391c470e4984ce34f76dcd4c58c09","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-23 19:14:17.000000000","message":"Patch Set 1:\n\nThis time it\u0027s going to be neutron-grenade-multinode.","accounts_in_message":[],"_revision_number":1},{"id":"07cfd2f5374c0bd100e2dc4243c9f4c3982628fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:29:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe6170f408a4565a7368df36fe5965b : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5420005fbd4f8ca3bf74edc2c10447 : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/638fa04a5af14ca7949baa992e4488a9 : SUCCESS in 9m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6af58d8f9dc4ee495d8ce0c102ce1de : SUCCESS in 9m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84e0c2c00b3940be843d640c1ec04b45 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ddd12129f8c486284ca77eef4af8ace : SUCCESS in 8m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e792d00ec3fb4aa79a8907558823c14a : SUCCESS in 12m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d358094f7bd04a7ebdac34180e65e4b6 : SUCCESS in 12m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4cf9335a98c4303bc43b192697c45f1 : SUCCESS in 8m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d592496cc82949c894d2b15ae6998144 : SUCCESS in 14m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87a63f04c4a3463c9d2bb1e246c1db9b : SUCCESS in 9m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/75037ab90f49431eb2b234d40cbe50df : SUCCESS in 33m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c26886df2fb444ab60b2c36f3df1fb0 : SUCCESS in 28m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3c222cfc3544421b9f76a7b92dfd73c3 : SUCCESS in 14m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/23246780798e46a5a9191507a9015ee1 : SUCCESS in 12m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba24e09fdc9941b0ac624e89337924dc : SUCCESS in 18m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2c068a7f41574df6bb26dbcb4807083f : SUCCESS in 6m 35s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/371350e1db7943c6a9c31ac5eb4e4838 : SUCCESS in 9m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/022f4b0fe0ac4ee09fd93fb11fbd0376 : SUCCESS in 6m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/82383f83474b431284bd5288104226f4 : SUCCESS in 13m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1c51473270344f93a4aec9809812926c : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bde4e2e1853e462aa6be7bd6cc5ba38d : SUCCESS in 1h 18m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1df6ea74f1f479fbc72506ff78a1b1a : FAILURE in 1h 12m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd66dcbc47c449d0901817bd78dab38c : SUCCESS in 1h 42m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/50cd7fee378a4ea39755f4c8f10debc4 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":1},{"id":"285351228d56e6c9ab07b88ec583818f7319c215","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-23 19:30:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"92768c9af8e5e7581c87e7c1f3da358e355bb342","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-23 20:55:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"80fd26f8e6485360ee71ffbc32edc5c4673d670f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-23 20:56:22.000000000","message":"Patch Set 1:\n\nFixes the regression/errors in the docs build for openstack/manila","accounts_in_message":[],"_revision_number":1},{"id":"6a127eae58cf198d2af696e443f314d66c244121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 21:13:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9191d0d5ad4a10902cc8e84e3146a3 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d031a4d1b1984ded8e25d63502a6ba11 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8025a0628ecb4225b5e960dd39059dfa : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b042ae4861a42f79ec11ae1b8c745d1 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31362093bdb0460f8c42937d5ca6f83a : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d40308e6479d4d8f8f11e55dd49d643f : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6114599664e4447aaf7719eb9f763f89 : SUCCESS in 3m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2d4ee1a820b440cda68eb21332ecd4cf : SUCCESS in 8m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c6b9252e929e4c07827dc9933785e532 : SUCCESS in 8m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbf21a254f6d459699635df7260f6253 : SUCCESS in 4m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/776398cbd69043a9a7ec0969215a8ca7 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/05328dbdfb944e2296ff792cd2e3c3a5 : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/16c19c5f07e447cda07b325d26287747 : SUCCESS in 17m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/09b2c0f5098c461c94de06a73166d347 : SUCCESS in 21m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b6fe239293c462ca4f34ace02ad4425 : SUCCESS in 11m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/407105a64fb94c7483fac6868de28e13 : SUCCESS in 20m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1617b212c6a420fb24017eb2bbe7f8a : SUCCESS in 17m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1e27743c8ee14467bc30d23b39c3b246 : SUCCESS in 4m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0aa038a8819f496d981d3bceb185aec4 : SUCCESS in 7m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe0eae2f78eb4094bf99ab85bbae3a3e : SUCCESS in 7m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/72a908c45f66412182997c463c28298b : SUCCESS in 10m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/46e11e0370464a53bb7db5fc6cc496fd : SUCCESS in 17m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ea03639c50f48faafa40731884ec394 : SUCCESS in 59m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3444910eb0304ab5aaf15a8528f28068 : SUCCESS in 1h 07m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05db219d2ec447f587feba4aec73623e : SUCCESS in 1h 37m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/202e19a21702438fb8d06860b610d114 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":1},{"id":"84c21878a6d851cc0854e91834a62973effeae6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 21:14:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1809f79d7819abf53595b3d5ab376ed0b06696df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 01:13:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bef6813cd5ca9586efb73a44c602c2ad64555a5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 01:13:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c3faf689494877a848d03854a2058b : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/309e4b34fec149519134b46200eb0ae9 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/144c89b4941d4884b61fddc0f1425182 : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f15bc0cbce6f45fe8a1eb2c264c95686 : SUCCESS in 6m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f306e480d5344ce18984858a78dd7b56 : SUCCESS in 6m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/143f2be74fc1452a906494e33521e8b6 : SUCCESS in 5m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c406aa00a9543b8a15882009a73d9eb : SUCCESS in 10m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b33b00b6b96b4d07aba6f2de095ee749 : SUCCESS in 10m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b3681cd4d8404ea9bcdfdccfd85860d3 : SUCCESS in 6m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9ff96c42adcd407fad673fdc9693551f : SUCCESS in 14m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d5bee4caa504d04b33692becc071657 : SUCCESS in 9m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/10d97aea9cec4e0e96c3ad4d326cdfa3 : SUCCESS in 14m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c405bc8931824dd79b58ad2def18e964 : SUCCESS in 28m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7bd1246d86b64859981b4fff5fd13def : SUCCESS in 14m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bfb293a167d248ff9a8abca8c994976d : SUCCESS in 14m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ea0a36ce7dd4621892c383a567d35da : SUCCESS in 23m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c5ef1b3e6bb34ad68167f2b205be9cbc : SUCCESS in 8m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bbd8cd29ae4646088158ebd605d9574a : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/406399e25227472b80fe038675de58c4 : SUCCESS in 7m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9431d8eb079c4ca69729b82215978e94 : SUCCESS in 11m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3d01559495044d6a855e82124d97349b : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b853d87e020a4541b0bf4f1114de42a2 : SUCCESS in 57m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/60f3a7701f134f4c8c96e2da2463524c : SUCCESS in 1h 02m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc726de3762049789cc32e144497ed38 : SUCCESS in 1h 19m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c1e141a6c07a4dfca079f2ffc905e860 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3ae48f3b58a5e07215c09089b8695a8e8b04dcbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 01:14:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2107c6b60eab4443a9522fdc5bb887f0 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f2835ccb942433db6e6d1046e9c227cdd821665","revisions":{"5f2835ccb942433db6e6d1046e9c227cdd821665":{"kind":"REWORK","_number":1,"created":"2020-03-20 20:03:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/714207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/714207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/714207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/714207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/714207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/714207/1"}}},"commit":{"parents":[{"commit":"02df513aa8c6de4469c5d5214b1dd2591aeebe1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-20 20:03:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-20 20:03:03.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 3.0.2","message":"update constraint for oslo.policy to new release 3.0.2\n\nChange-Id: I9922429bc3d9de60d1bd427082c5cc37c88db9e4\nmeta:version: 3.0.2\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I6c1d8412aee4a695187892f8568291c1203c9116\nmeta:release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f2835ccb942433db6e6d1046e9c227cdd821665"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f2835ccb942433db6e6d1046e9c227cdd821665"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
