)]}'
{"id":"openstack%2Frequirements~714243","triplet_id":"openstack%2Frequirements~master~I7b8018ab86b2c440b9850a73a5270f9c3c5d251e","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I7b8018ab86b2c440b9850a73a5270f9c3c5d251e","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-03-21 06:13:51.000000000","updated":"2020-03-25 05:03:59.000000000","submitted":"2020-03-25 05:02:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714243-1585112542578-354c625b","meta_rev_id":"bb54338eacd3cfc57b5fb972430376656dd8a8a8","_number":714243,"virtual_id_number":714243,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 05:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-24 18:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-24 18:42:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 18:42:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-25 05:02:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b79a48f8e67743ecdc640a167a3f8e214943b858","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-21 06:13:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"471688a6b5cfae8c982e89282f4ad5a95da8dbb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 07:53:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2198942caa8a4e37b0f4e13a683e51b3 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f83812e48947f9b9e5f85748fd42a6 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c70b18ffebf64108a7dbc3c866539f2a : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/677a152543804cd5b5071436f45a243c : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fdbf5a4cbba4320b19d17a55eedc954 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55aba95804b34a98af3107a9c17da38a : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1afc2078b9f74d6ba1abadd9b687abd0 : SUCCESS in 5m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7a54676d636740de9eed4971b391b853 : FAILURE in 12m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d757e1de969548d99ac36005c8d230f1 : SUCCESS in 14m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c0cc8204b3134b04ad5812233b447311 : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/265f447cd24641c9b40f0fa15e94648f : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fcff8de4816b423c88a249396c11097b : SUCCESS in 10m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4c91d2facc3f49ce8f37001e67debcaf : SUCCESS in 12m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7563c54ed16d4f36a85108cd6910e448 : FAILURE in 31m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fb0cccf7c2ee4490a8ccd2d3463989fe : FAILURE in 13m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0796656409cd42d5a560f6d8c33304cb : FAILURE in 14m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f1cc123827441c882f5a4d68990ed4f : SUCCESS in 19m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9c70f11072ce4d4a9108f278db319337 : SUCCESS in 5m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/28a1ab060d0a416281e0a1de4dafe189 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d168a13ec01422f992a6679c84e1219 : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/46cdec56a5d045018d3615d03da4998a : SUCCESS in 10m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/09a6b58621ee491e90895c6ad7d532e3 : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f019926a51ce4955b86a2d8f900fdbd1 : FAILURE in 33m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e33234d300234a8983ea7e75137fdd91 : FAILURE in 58m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0288089d8e84db89ea3f22e155291f0 : FAILURE in 26m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/beaac3b62f284deba8090e0461c93053 : FAILURE in 1h 05m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0a53b6f4cc8ce5d7220f55ddbdf2277f884decef","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-22 06:14:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1260f764952397bf406ba67262134fa3405158b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 07:30:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffaedeeb52f14d0baf675f19e68530d5 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3040484f14d8414498eae0e5863d724f : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/177391db25a04e46bcb14ff085c6afd0 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9098a4ebbdc4ad3afcf6644b4595df7 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4734a1735b84113bff96c10c59cafb6 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/680233b28caf48e88ab8557d34332ca9 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59310f99470c42d48e4d0e91b3ecbab7 : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7099d795ce9d417e9456986cabeddcd3 : FAILURE in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cb0ef0d3ce2849058a73efae20b0bec0 : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fed0003c4a5a410facb9c1c08f2deaa9 : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/480229d83b264537afed29ef868a9cea : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f64af1962d54372a2d4ecb065fea17b : SUCCESS in 8m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0ec58f461f4b4f9286a8f6187abbec43 : SUCCESS in 14m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/810af36865b14b388bef6950a2d238de : FAILURE in 35m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/10b5c9ed369d4962915e3020617bf867 : FAILURE in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/db964c7ac4424e988001ccdaf76b6421 : FAILURE in 18m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c522705973724c0ba13ab97c5b2c20e9 : SUCCESS in 17m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/79426910ca16474c816f9a435f679322 : SUCCESS in 5m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0321aaefe9c348cdbafe3cdaea82e432 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1059ff2058d74582b4f1ff447d14c972 : SUCCESS in 6m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac6c15e94368440794862d28b637cb19 : SUCCESS in 9m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3ce82f20e66f4d71bd4b540a95332621 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9b7f672c31c4c0dad7c5557a958628f : FAILURE in 35m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3dd457ac5cd24fe88342af5ba673e40e : FAILURE in 1h 07m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbdf881754f34113b5a2bdbcb4bdc7df : FAILURE in 30m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8f5a1de4dfee4ce7ac84947c837dc9f6 : FAILURE in 54m 13s","accounts_in_message":[],"_revision_number":2},{"id":"c12fc5aaca7b777c1b350f1ba03a1bfc4f9a2196","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-23 06:17:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c89935c5cc01d70605528f182482a8f68f988b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 07:48:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80df922d8f26460eb79eb341765d6524 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdbbf8a721e54a479b5c327a0b7724cb : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03bdf64eee674156935b3df096209751 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2a05ef5cdc844da8cc846e20988994d : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a48552a3bd0041f2a8872cfb219c67b7 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/776f9a1f8ea74ebc85722eab29b418e2 : SUCCESS in 7m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0a3401758d4483a9814534607b00a72 : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3262ac9b84b24ee0b8aa734e0c890b5b : FAILURE in 9m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0e790f0824c846ec8e8ca4dd660ee128 : SUCCESS in 8m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5b121d48cd5d40e39192f1ea8899ffa8 : SUCCESS in 7m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a336139b19ab4dc2b10cbfc4cacdb6e3 : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76840940096548a9b53866661eb0284a : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/96ebe90558ac4284a8b350363c50df93 : SUCCESS in 17m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bd0117ac0eb64e1785427034843ee798 : FAILURE in 25m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/021b85cf538548c4bf89f3d21ebaa1de : FAILURE in 16m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2bb7f194e8534a2597843353fffbdb8f : FAILURE in 19m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff43ed88f01645fdacd671c8568a8034 : SUCCESS in 18m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cd07c9751a5f46b8b29fe0b5a945458e : SUCCESS in 6m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dd35dfc87bdb4016a55f847ad5a33fe8 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb1f71ca0fa44c20a61fe2376cc4f2fc : SUCCESS in 7m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3640c42a73be4bd3ac9fff5b75fcaec8 : SUCCESS in 12m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1313854d3ccf4ca592bae9d8e6cce8fd : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec8e4dc50374492ca1678341d6e4fd51 : FAILURE in 36m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8cbddafb851c489a9368ec331a9ed33c : FAILURE in 1h 06m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54437bf0473f409585795148935bdbe1 : FAILURE in 27m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2c54107e434e4605af9c1ca3db41e9a6 : FAILURE in 1h 07m 57s","accounts_in_message":[],"_revision_number":3},{"id":"11426e28eb2d37ace64f0c476840beeb743065f0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-24 06:20:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8473030acd793f502cb81d62cd9b1dd637d002a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 07:54:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a377e582214429c90adc5c530fa8fae : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1260b20fba44524b6ffd0791b10b63d : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea03a9175c8749f9a34f02265cc328b0 : SUCCESS in 6m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67d94679d20e48d5aacfce779f21edc1 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9db91a4b92444cdacb08b094d6ae11c : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9b22753c15545948036074cc2723f7e : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9dbc6fc6c5941da97ae04ac1bbed785 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39c85ef0872947429425cd2b5683b64b : FAILURE in 12m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8edc0a870dbc41c39c89dd218b8a199a : SUCCESS in 9m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/155b4b12ebfe4bb386fe6612f50c3be9 : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/51d099aa4f124721b717605693482a09 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e7f68d8c00ca41d5801016eeb968788a : SUCCESS in 8m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6678eb40395146259d3bf1022f9ba7ce : SUCCESS in 21m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a8190d5568eb402cb6a33346c87f28f4 : FAILURE in 22m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8274be997b0a4c30886fb6eca5f0d676 : FAILURE in 14m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7f7f2d00398f4286a2f813c81ba50e78 : FAILURE in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de6ccdcce1c547e9a790896829403c4b : SUCCESS in 16m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c62e9d505804474a8e5aff10837e47bd : SUCCESS in 6m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f1372ff5222f4a34ae9b90e5dee9c7eb : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/92538de36e4f4445b3fca957a5adc59e : SUCCESS in 8m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c6dec3446f484c928de8b669c1cb8776 : SUCCESS in 10m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9e55f3c5cdf548619a46c972fc2c9aa6 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e397a421166494780ca284972c88f1f : FAILURE in 34m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/abcde159890e4a4497d6a463a128bb23 : FAILURE in 55m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8d06517fb0a40d1b373e7fd7686ffe2 : FAILURE in 31m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a6af7f1569654bcaa766d4e153ac984d : FAILURE in 58m 05s","accounts_in_message":[],"_revision_number":4},{"id":"4a7fed348259f429f24e9bbba67bceaed5f5bddb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-24 15:29:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3f41a314308b54c777d4657b9db45f25db464b7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-24 15:30:01.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":5},{"id":"ce6afb5ffa495e8d9e22091f3ae5a47cda044fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 18:34:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9aeb7bd2df74c97b55e36db3653ff30 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ac8bb058194ea4bb1a35e64201bf2f : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48105ab347c844aab163cc8dd6b3aa15 : SUCCESS in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd8e21edcf514a6eb9a68001ee78765e : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d35c29576cc9497e906df55c52a8cfb5 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/107cc7ab32a64dc0b54e6b51187b5fee : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/70d7a612e78f447b9335a13318e379c7 : SUCCESS in 5m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/452ac0cc2f784c3ead29414573068ca4 : SUCCESS in 12m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e9e566e1366d41b3a8d02024fba4fb20 : SUCCESS in 10m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4724895d140d4eabbf80fa8428f26382 : SUCCESS in 6m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d79921f097034e94a9c96fd0d3ec3509 : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d05e678848974fdfa3db33b9063ad7b2 : SUCCESS in 9m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7f40aa03069f412cbbebb02912f2614e : SUCCESS in 15m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eb85ccda13bb4040bc59fafccefab162 : SUCCESS in 29m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d5757ffbf90b45d5948cb8c014924d3e : SUCCESS in 13m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/75897a1d0e4a47189f6e2fa1acd1a7a9 : SUCCESS in 15m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b52c3e2570748bdb9dd482cb437018f : SUCCESS in 17m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b334ef1486b47d79072723c5456b982 : SUCCESS in 5m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5ddefa20f7e8408680def63583deb641 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67333b3402f34cba81c907dc41a614b0 : SUCCESS in 8m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1a661af1af8e40578c7ce2f37c196e21 : SUCCESS in 10m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3ee4ad69928e4603ac90067ca1a26fd0 : SUCCESS in 14m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d79f7819066f4ede89bcc9fb6572f2f7 : SUCCESS in 59m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6d4803253bd4ecbb6104c51405c3db1 : SUCCESS in 1h 13m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e3f975118dd4d3189c00296330337ad : SUCCESS in 1h 32m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d2f1faae0a55407d879cc5147b52f6fa : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":5},{"id":"f2a5cff42f5c407b92d8f1b71b3ad2ec5eb45488","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-24 18:42:45.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9da34d96fb308221c25becc5085cb712850eaa30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 18:43:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d2b6c4fdb18b0431f053dd330c4d157269242901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 21:00:09.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e17b65549d1d422fa0d0299496efd9ca : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a0a552c935d41a89d2cdeec64423b06 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ea3e844471c437b968aae33e4c67502 : SUCCESS in 8m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28831596c76e4c6cac623924a3d22359 : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f05b8e5d41924fc5857749f3869761cc : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dffe078c6dca4ce3b7beb2b99f7ce987 : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1865d65df5b246db836344098ead9cd9 : SUCCESS in 10m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ba48272f04ff4d2f97bf459b38592b93 : SUCCESS in 11m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9bedbd64ed6d4349903c4688e53ac43e : SUCCESS in 6m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a2348c50324240109cb353d76d189e99 : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c88b655f51fc476cb8001691307a1b9a : SUCCESS in 8m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7893c372d8a14fd8a03a29301efdfc83 : SUCCESS in 17m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0dea9527965643d98da142048718688c : SUCCESS in 26m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cda2f9e1ba604ebe96f33c74025adbc0 : SUCCESS in 14m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1b1def6ae7874b00a7bc27c910382a91 : SUCCESS in 15m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/904aa9ef485148a79a38b081f8fe9382 : SUCCESS in 17m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d71324a123214ef7836ddbb893fb36f3 : SUCCESS in 6m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/11c65f676d324afb840079f64dedd260 : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d2042215653c45f78043b458f709f375 : SUCCESS in 8m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/615c2eb394aa4d46bd602e9a14eaaffa : SUCCESS in 15m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4d867c1f422445f1836d8f95657fe053 : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4284ab0c3034874bef39a881c474454 : SUCCESS in 1h 01m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29d5e501ee79428c91bb2b806119cf27 : FAILURE in 1h 06m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75ffc82bd5d44b808d67a63ede2b8b8c : SUCCESS in 1h 31m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a373bcc84d9e4a70a926e1e49b13e27a : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":5},{"id":"1474d381f09f9809da454a3d20c6bc7c7874dd5e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-24 21:27:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4d24b8ecc403d20cc1f1ccada3a316b0a34413ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 22:56:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e1b0529d99146298bf69fde86ee8da8 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7fad8b35fb44667938c20d9e694c5a1 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04d1f1b4b6fc4bc0a2a8957f1629dfb7 : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/147f720dc3934ff2b61cad17b69ce7ce : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a500cb948dea41edbf3b926b5eeae1a4 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d522de86bcd414d99ce1740eccd965d : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3e62a9b806544dcadc8ad0d1e2b0eb1 : SUCCESS in 5m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/07ad999a7fa94f00b16311040a044415 : SUCCESS in 7m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a4bf8a56d24c4da487cba4caec474a9a : SUCCESS in 9m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f36fb567e5a74e23ad554b98aa040f7d : SUCCESS in 6m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b443a88ab11948bc8823d556e678b21e : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0b477e7dbee4bf4b88ac93b1f1688fd : SUCCESS in 8m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bab4d26714374354840f82d35f8b7c0c : SUCCESS in 16m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1c07d80907bd4a448d7f0e1ac219f99e : SUCCESS in 35m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6d189d20b8644c54bfaa755facf4713d : SUCCESS in 12m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8c6ceb9bd311481cab0652d175ce7dd1 : SUCCESS in 14m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44c5ac73595b4e56b65d3106489e39f9 : SUCCESS in 19m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/17386c3b7a044218bbd35a6deb1b074b : SUCCESS in 4m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b4dfdac4974e41feb17a9ffcb612af7c : SUCCESS in 7m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eaa2e5b052234cd3b2db5638262909f5 : SUCCESS in 9m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/df13b55b2f8a43109eb86e42fb5df604 : SUCCESS in 10m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0a1446ae9bc84591a23389ca54680d5b : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66c08cc58e994bbeb1337199565149fd : SUCCESS in 1h 00m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b54f961ccd814bc189b6af0322134c29 : SUCCESS in 1h 03m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0c79921f7b74d29a3160dd8d6fd052f : SUCCESS in 1h 23m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/23686cad9f114b5dbd39a2573f8c7956 : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":5},{"id":"e62b2751795b8c2f0514638b2b2bb350e98f5833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 22:56:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8dad5ce7d31ce0156efcf5f8839852af609459e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 05:02:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae4fe5933cc45bc97f3f6c51acadb47 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9484665524c044ad800bf37bc56088be : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/733c65cd0b5740c49cbcc41f996370cc : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b8334e381054da2a71d9a1921aeeeac : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/695057cea0304f86b3fe33fd2f5a07f9 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ea3ca174ec646dbb0af035c0e701978 : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a764d440e1894aafb68b5b3d41374595 : SUCCESS in 7m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3d25f50303a942ed81a05343253be36a : SUCCESS in 8m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/52e51bf8437f4698b320169802ac4806 : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4b783c75fcef48ce8d65b8af25d6d918 : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8495ae2e82a84be793578bf5c35e936a : SUCCESS in 7m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4774376a6aaf4cd49cb606377131a36b : SUCCESS in 11m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3c52aa389fe949ba9e4a999de0e0c307 : SUCCESS in 28m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/76579b6a84fe4c8f9ac81e3a5dc6672d : SUCCESS in 16m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c0868ac2ecc944369b933a10fb8921f3 : SUCCESS in 19m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fdbdf15f3f9f4cf29166d134144fd282 : SUCCESS in 19m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/947513ad1347454aa35969ed3c2360b6 : SUCCESS in 4m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bfaa6ea0d0ee454aa3b6176f15153da8 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3d2ae90caa74c9eb723ec51d58bba6f : SUCCESS in 7m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/44facf6524a84b8aadc3c9a4aa7b02c7 : SUCCESS in 10m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/519c9bb0e6d6492e98032cd18dcf522d : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df380a5bfdaf4adbbd8d37382184ce29 : SUCCESS in 1h 07m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0474ecc3d0f4f7ebb417a48323fb256 : SUCCESS in 1h 17m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e3d73ae62ee43658f46cb6bfec167e9 : SUCCESS in 1h 31m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c27c6e02993b48c0bceb26a38645b0fb : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":5},{"id":"b33137f22a75a5b2d52f79258c1e61ba277d4606","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 05:02:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"bb54338eacd3cfc57b5fb972430376656dd8a8a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 05:03:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030708234f9b4b78ac21c0491f2d8a0f : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d418df0795e1500a9180226801c2633c877bcfcc","revisions":{"8ab62d9159d994d7d5002cd67f9000ffacf66a7b":{"kind":"REWORK","_number":1,"created":"2020-03-21 06:13:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/714243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/714243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/714243/1"}}},"commit":{"parents":[{"commit":"02df513aa8c6de4469c5d5214b1dd2591aeebe1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-21 06:13:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-21 06:13:49.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b8018ab86b2c440b9850a73a5270f9c3c5d251e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ab62d9159d994d7d5002cd67f9000ffacf66a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ab62d9159d994d7d5002cd67f9000ffacf66a7b"}]},"branch":"refs/heads/master"},"a5be8eeb7b489c8db5c6dc483373e4a52830d40e":{"kind":"REWORK","_number":2,"created":"2020-03-22 06:14:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/714243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/714243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/714243/2"}}},"commit":{"parents":[{"commit":"02df513aa8c6de4469c5d5214b1dd2591aeebe1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-22 06:14:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-22 06:14:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b8018ab86b2c440b9850a73a5270f9c3c5d251e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5be8eeb7b489c8db5c6dc483373e4a52830d40e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5be8eeb7b489c8db5c6dc483373e4a52830d40e"}]},"branch":"refs/heads/master"},"48eebd96f9e4ed7268f08f6c0316df3b66e10f4a":{"kind":"REWORK","_number":3,"created":"2020-03-23 06:17:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/714243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/714243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/714243/3"}}},"commit":{"parents":[{"commit":"02df513aa8c6de4469c5d5214b1dd2591aeebe1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-23 06:17:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-23 06:17:28.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b8018ab86b2c440b9850a73a5270f9c3c5d251e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48eebd96f9e4ed7268f08f6c0316df3b66e10f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48eebd96f9e4ed7268f08f6c0316df3b66e10f4a"}]},"branch":"refs/heads/master"},"ab2e984dc0ba1f7a2ab279ea49060835a38d00fc":{"kind":"REWORK","_number":4,"created":"2020-03-24 06:20:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/714243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/714243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/714243/4"}}},"commit":{"parents":[{"commit":"6f269f470f26b2ce2588aaf1df532d6154f6a524","subject":"Merge \"Soften the py3-only warning check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f269f470f26b2ce2588aaf1df532d6154f6a524"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 06:20:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 06:20:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b8018ab86b2c440b9850a73a5270f9c3c5d251e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab2e984dc0ba1f7a2ab279ea49060835a38d00fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab2e984dc0ba1f7a2ab279ea49060835a38d00fc"}]},"branch":"refs/heads/master"},"d418df0795e1500a9180226801c2633c877bcfcc":{"kind":"REWORK","_number":5,"created":"2020-03-24 15:29:50.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/43/714243/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/714243/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/714243/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/714243/5"}}},"commit":{"parents":[{"commit":"6f269f470f26b2ce2588aaf1df532d6154f6a524","subject":"Merge \"Soften the py3-only warning check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f269f470f26b2ce2588aaf1df532d6154f6a524"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 06:20:27.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-03-24 15:29:12.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nhold back mock for nova/neutron failures\nhold back docutils because boto likes to cap things\nhold back msgpack because salt likes to cap things\n\nChange-Id: I7b8018ab86b2c440b9850a73a5270f9c3c5d251e\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d418df0795e1500a9180226801c2633c877bcfcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d418df0795e1500a9180226801c2633c877bcfcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
