)]}'
{"id":"openstack%2Frequirements~714490","triplet_id":"openstack%2Frequirements~master~I1d268b17935ef4b2977e462155b86633b1fe1442","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1d268b17935ef4b2977e462155b86633b1fe1442","subject":"update constraint for openstackdocstheme to new release 2.0.1","status":"MERGED","created":"2020-03-23 16:37:15.000000000","updated":"2020-03-25 18:55:15.000000000","submitted":"2020-03-25 18:53:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714490-1585162386893-fe09d30a","meta_rev_id":"612e642d326c5eb081faaa49f5355c4b4f2a7e21","_number":714490,"virtual_id_number":714490,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 18:53:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-23 16:40:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-25 15:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-25 15:45:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 16:40:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-25 15:45:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-25 18:53:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afdf1a0817fc0d919e7a912190ecd1157278af62","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-23 16:37:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b12f12de76a2148529366705aef56d3d4fd4db7b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-23 16:40:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"914d792fc82bb2df0e6afd5181ef8e894c1081b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:53:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e738b0633c416b856b1436b550f246 : FAILURE in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5d7562a8634e4b913edbb4f20a6ee6 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe365c2a22dd4f62bf1c80fb35a8cc77 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e578a07a3b744c1ab474a904ea7ce1ab : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37de49e897e7488f8f0151653195b19a : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7419fce485544a3ea133b41a484e80fa : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fab1d834ebe841a1be77f1fb9e896b3a : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0981b2c3d8684183832853e25c93c5d4 : SUCCESS in 7m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2dc6f11b4bd74d9fb34486344781fc0b : SUCCESS in 9m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/beec4119c1644ad9ba7b638e2ed8db97 : SUCCESS in 5m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/41c570826dc4461e9d312c955fa94e45 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd0271bded7c4054b2f2ddbc299b5e03 : SUCCESS in 11m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9eb3c2854eff4ae09f7fd692b86cb683 : SUCCESS in 23m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6576cd4b9103494f937321cccb103e1c : SUCCESS in 25m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2431b33f967b4e49a61a8782f3cc0a9c : SUCCESS in 11m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c395f53bfe694ca2afce602d3750cbbb : SUCCESS in 13m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c085224c21044eeae5131b245a396a0 : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/72386efffe424d1a8fab3f226d21e932 : SUCCESS in 5m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f065245bffc4482bac6134902d1d91be : SUCCESS in 7m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6c85aa5b17b45208ff507bb0e4becfb : FAILURE in 4m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ec3c144878b5435cb505f734c1019594 : SUCCESS in 10m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c2298cdbb7b54f0fac4c2f89de84bd11 : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2198846334fd4f5580592196483c5dbd : SUCCESS in 58m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/11d727e1c44d40baa18b31f2668628c6 : SUCCESS in 59m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bac5ea996af449109f5c31a14baa4792 : SUCCESS in 1h 14m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/25ff523c4afe4948847c16a28bf1abaa : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f2cc24e72929935791d28f986a7485f87c8c9cad","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-24 15:25:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88554b09aa414a37f2d20b66d5f4648c025daa74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 17:34:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1cc9542fef24d02b935f26819939593 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1518761e83684efba9788c6f290cb4bb : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3483ea51898e4b40a3c8fb209d09441e : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10b75f063d4c4182ba8fc5ac8c670fe5 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e85106b378a144e2ae7418ef48f6573f : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74d1f40f12cc453ea48b201538c54762 : SUCCESS in 7m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/219079d0329d496f82f9b8d603b72d3e : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4ff7bfe5590a43beac45ade8f96ec854 : SUCCESS in 8m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd80cf4db6cf4f4395397b4079280c05 : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab7f4cd9c90c46599f13ca2f63d0d246 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1697066bf82044ec93fd8f601a5e916e : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7d4959539a845ebb3cf105aedbb5bdd : SUCCESS in 9m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2fe2e3eb473c40e4aff69a34ef586eb7 : SUCCESS in 14m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7eba976276454802b53581dc721da636 : SUCCESS in 27m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c213ec69afd544a698551611d2bb174a : SUCCESS in 13m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/571f712712604a438da8b42cdb5fa35f : SUCCESS in 14m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5d93ae7b29b42fda0e42d5f44c6ebb6 : SUCCESS in 17m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b4e61005824747708d16fe7cc4cf28af : SUCCESS in 7m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/02ec5a5da29d45f4a40b91f2df993382 : SUCCESS in 7m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6bc3a9f34a14621b39d3339a5e53a78 : SUCCESS in 9m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/355fe07dcde44e4ba13f6c49171dba04 : SUCCESS in 13m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c2e34bb9de5c437485b5099189ba9e50 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/def2696bb5214aaaa5a8da2dc80905e7 : SUCCESS in 1h 03m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f2b9947464d4da1bec959d41488167d : SUCCESS in 1h 10m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85d55454fb4f4d3dafb7106db371a7a4 : SUCCESS in 1h 22m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/af6c7bf9743b441186bb4e9cc210438a : FAILURE in 53m 50s","accounts_in_message":[],"_revision_number":1},{"id":"da9f235951eb1affc0241638974cd1bd2e1fea26","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-24 17:43:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6bd0452b07d1499f55ce20174f3f5bb17fe0f988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 19:09:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60c733f91974785a0739bb375e2f989 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/673ae5a916a54085b0cf8bdec9f42cca : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/defa51b39af34467aa751ace4dfa770f : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfb8728288f74ea08ce184f1f8e2c34e : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87406f177cff4cce88a10ad1a6a7f9dd : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f4870d58541423db745f28209b476da : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c28183ea30df4d3bb9921b09ff6a83f8 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8dff283cfed14f058d4ef517a20d9243 : SUCCESS in 7m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d4fb8188390241808da3f1b2a0c45156 : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/74e8c687e5d64adf94dac7176c76a425 : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b9d1b66ff1e94573907c7bde37793306 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aab17888f18244559fe479d09a06fa5b : SUCCESS in 10m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9b7152f440ff4735b09e5dccba2e6fc2 : SUCCESS in 13m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c644477b81864fec9cfdf935d7384067 : SUCCESS in 30m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/67261a0d61b0490e9959b2561dcea8f6 : SUCCESS in 13m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3ae4c9b4099f4ccbaca5353aca95ed5d : SUCCESS in 14m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae1db7fa135d429ca61cc780f752c9bd : SUCCESS in 16m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/060ad4a494cc4e5498afec12a39c92f0 : SUCCESS in 8m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/13c1f20ccb7f484885f25e9579ce2425 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84131aa70195493cb8768b6f3aaeddd7 : SUCCESS in 7m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d6e658c78c28417287f02f800b6666f1 : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a5ebe0612d2f447b9b356bb1826871b2 : SUCCESS in 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d34dd5b4909e4d0d86be53214f6f5926 : SUCCESS in 59m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bfa85a2e51e42369283821389d6afd9 : SUCCESS in 1h 08m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4daf702951fb4ef094d0b1e52cf15ca3 : SUCCESS in 1h 15m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0b67ff0e10624f6385fdb569c1f7d796 : FAILURE in 1h 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3042d0ac4d2753892471086ec2ccc9cc90a400c6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-25 06:36:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a10ea59d8d37e9822108072a5e76ab7c31fd4d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 08:40:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/584ead6a60ef411080a2965d7ae295bb : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f699d8d30246c6a99926ac392286db : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38c50598a6d74207a88b61585e32d9ed : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f29212f5024f4f5eadb61755d73b0d23 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42dc99fd78bf468ab6678d8f6bb40911 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4eac9c08610849849fae029286a0ea0c : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9cbca240c03c44c7912126a9f48ddf39 : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65bf460362fb47048fdd6216024249b6 : SUCCESS in 11m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e709c1151c5741a396a328eacc70518f : SUCCESS in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4efcbb9560c744818753810c57010e9e : SUCCESS in 6m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3aff2876e2764e478ce1edb9df4e346b : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e3f2ee18c244282b427d447f4d9a49c : SUCCESS in 8m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bd7ea7061ccc4d9f871c5b4e988f647b : SUCCESS in 27m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/be33138cc50f45c88b4b6e827da2667c : SUCCESS in 25m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be184ea0c6de4035a7815538f639e0fd : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d1ff9638df8f4031adfaf16ca14244f9 : SUCCESS in 13m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/289c144bf89548099de57dbda8449b10 : SUCCESS in 16m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e15b391174c844daaf6ff119de8fc92e : SUCCESS in 6m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1aad49a2786c4f83b293895f4ce9da35 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89761633fac7446c8b465889c7dac1df : SUCCESS in 8m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5a42e0fb208a4fd989d0a2e207c2a53c : SUCCESS in 13m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/91b4693b382747cb9152d267eebfe223 : SUCCESS in 14m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/102c2acdb631421cb1133ed213c491c9 : SUCCESS in 55m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/42f9151cba464505b41a046ead0e1918 : SUCCESS in 1h 15m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77ead7038a8842a0b4e7f877b6c0779d : SUCCESS in 1h 42m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a0c78bb2fb4441db9a317bfa1b3b4588 : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e51dbd8f6fc2d05a951d68721c591f17d1098150","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-25 15:45:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33b251dc256413f79dc0c844a099ec9f3c745e81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 15:45:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdcf3fdd8043dccafd641001b6556341c4465f67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 18:53:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cd60088b362f1cb3cb3d68e9f8be06e62f1922cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 18:53:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ce763554e444faaeb8050c0375d883 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5620da31d0fd4c2aacaabe88b0939aa1 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6abfd7f20be54e82bc4379789b893d35 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d93b882146a4e5fb515ebdd6eb01010 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a19a7fed5ce4bd68f55bbf40c37d052 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed07bb83c8b045fe842ba7a15caf72a7 : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2219ace7d9445198271133c65a40e56 : SUCCESS in 8m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d3c47ca91dc44a496478e57c2ed4b32 : SUCCESS in 10m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9ad3a189f7cb4386b77feedb9d8523ef : SUCCESS in 5m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5ea73344cbb74eff956955a1be74fdfa : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/342aa0c70d5547ed9651db0b0df18f22 : SUCCESS in 9m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/258bd381b6eb45aea1bedd3aed8dbbbe : SUCCESS in 21m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/25555aae6f6c49739a0651260788eda7 : SUCCESS in 23m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bfe2580bc51a499397d922539b38ce65 : SUCCESS in 14m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c08e7c2691f4418c9f97adaca9a3045e : SUCCESS in 25m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e237322b2a242bfaf1937a1fd6e47e6 : SUCCESS in 17m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4d64c152a43a4aa39385c83d99fbde4d : SUCCESS in 6m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bdc7d698b6064add9f5208de350e856e : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b9fba010e2ce47999177442d2ae823bc : SUCCESS in 7m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/da8c0fc1950343bf90874cd592780cfc : SUCCESS in 10m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/25320fc509454c9b9dc691e29342dbf5 : SUCCESS in 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5867e6d11174b9992cb9022bc9396f3 : SUCCESS in 1h 00m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b00b771afb0f4429952c3733c4508fdb : SUCCESS in 1h 04m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd632a13c366466dba34fc1502352951 : SUCCESS in 1h 34m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8343c75105d5489faa5549878d657ce1 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5e515f5e02b127d1d113ed11fb53f31e32b008c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 18:55:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42bd5548c7a44b62bbfa58ee7c45c547 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a93a73bd4d6211c384ede95f49968e726ca30316","revisions":{"a93a73bd4d6211c384ede95f49968e726ca30316":{"kind":"REWORK","_number":1,"created":"2020-03-23 16:37:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/714490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/714490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/714490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/714490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/714490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/714490/1"}}},"commit":{"parents":[{"commit":"02df513aa8c6de4469c5d5214b1dd2591aeebe1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-23 16:37:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-23 16:37:12.000000000","tz":0},"subject":"update constraint for openstackdocstheme to new release 2.0.1","message":"update constraint for openstackdocstheme to new release 2.0.1\n\nChange-Id: I1d268b17935ef4b2977e462155b86633b1fe1442\nmeta:version: 2.0.1\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Commit: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Change-Id: I20722633fedf28a31320867cd7ab9732297c91a8\nmeta:release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a93a73bd4d6211c384ede95f49968e726ca30316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a93a73bd4d6211c384ede95f49968e726ca30316"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
