)]}'
{"id":"openstack%2Frequirements~714522","triplet_id":"openstack%2Frequirements~master~I4c770fab0366f70e7e01920498498a3b79bfc423","project":"openstack/requirements","branch":"master","topic":"checkcheck","hashtags":[],"change_id":"I4c770fab0366f70e7e01920498498a3b79bfc423","subject":"Soften the py3-only warning check","status":"MERGED","created":"2020-03-23 19:45:59.000000000","updated":"2020-03-24 01:15:46.000000000","submitted":"2020-03-24 01:13:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714522-1585012409702-5004bf0a","meta_rev_id":"c0a5e9f32868d43a427ad7df754bb1215c332093","_number":714522,"virtual_id_number":714522,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-03-23 21:26:18.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 01:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-23 19:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-23 20:30:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-23 19:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-03-23 19:51:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 19:47:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-03-23 19:51:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-23 20:30:15.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-03-24 01:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55c01824d7b9df57d4cee0991cea60fffb65a7f8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-23 19:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91c923eb452093a5fe51666db984d32bcda54e12","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-23 19:46:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4b6d75e5c2a53abdfaa6b7b3592c9b130db130a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-03-23 19:47:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5b4a08a826d0e875c523db2799f1bf4b6faf2e23","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-23 19:51:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5a6d7c9d7e5ccad59682d5c34a8fd41cb4d85f54","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-23 20:30:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"08f172c203caa09c700907d7182e5d8d944b8ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 21:15:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6cfefcc68142aa9eb1b1f2f2593b21 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fecd655b44aa40779dfd59cca2ba5244 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd321d96fe194cafad7a177e33891b17 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53aab3ad4a534f03bf71b73b8de20b8a : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae6584ca36d84fcea4543a5f0cf300ad : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99296c74b9514d4c84ec8ddcaa6c61ed : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19369dbbf6a0415299936021984821f4 : SUCCESS in 45m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5e154081b9446feab1811c1d2160a8c : SUCCESS in 1h 05m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df2fe84a7a86451b8bb1aed1be520199 : SUCCESS in 1h 27m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/929425127d2b4bd6a6885139de09d976 : FAILURE in 1h 02m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ef6c459a5b222f0c75761e567a96af3d8637636e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-23 21:26:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e8d4ad598087c1fe6c268bb7f814186aec8919f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 23:07:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e15adbd036d46c392513d5d2d5e4909 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf2dbf2e97a4b78a42c758543139be5 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d58278940d6642d7a975370bb43bd0a3 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1877bdea041464f90fb10fa20be5b91 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78121e3a319a4371b543dc16d5c4cdec : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec76d653c1bf4ee59b06d92ebb88f797 : SUCCESS in 3m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5717e9852b054767bbe24f2c8b58d85c : SUCCESS in 54m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/24b68bbfc98741f98fb1c7d8910f8c7d : SUCCESS in 1h 12m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8876f2b5459445bd8f03abc8290b0ac7 : SUCCESS in 1h 30m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9935615c88e14661ae0894087d4b74ad : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":2},{"id":"c017888421a96dd5d0d5ae62f8bb08c6587a9eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 23:08:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2603cc8239c5741969e7540ab0db808eab5c0d85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 01:13:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50f69c18b444447bb8c6d691147e7c6 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0f08297f244a93b58aba1468804680 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e414680781042229e86d6f7186e8062 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f1862605a3846f2abbeede64bfc7a8e : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83623621c74d4e5aa8c464625f81be7b : SUCCESS in 6m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed0b3f06e7434800b29db570550fc7ea : SUCCESS in 1h 09m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/745097769c534b61ada56c981b25842e : SUCCESS in 1h 06m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbc38ee487074adcadbe059dd81b19a8 : SUCCESS in 1h 36m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6d27e0f519884e5184f7ac89f4e23608 : SUCCESS in 1h 20m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5319f16dc3c1239d508f698a85b88d9a6afb97fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 01:13:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f69a50c5437f4a1566fe00a9955bc7c02705e99e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 01:15:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8900c0effc4e1c924cd1464e1163c8 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb128fca0294f047d0071f4feb6564930c159d6a","revisions":{"1c6ec27939d7a227a0882f35ba3ebdb7c2d97f01":{"kind":"REWORK","_number":1,"created":"2020-03-23 19:45:59.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/22/714522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/714522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/714522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/714522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/714522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/714522/1"}}},"commit":{"parents":[{"commit":"02df513aa8c6de4469c5d5214b1dd2591aeebe1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-23 19:45:52.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-23 19:45:52.000000000","tz":-300},"subject":"Soften the py3-only warning check","message":"Soften the py3-only warning check\n\nChange-Id: I4c770fab0366f70e7e01920498498a3b79bfc423\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c6ec27939d7a227a0882f35ba3ebdb7c2d97f01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c6ec27939d7a227a0882f35ba3ebdb7c2d97f01"}]},"branch":"refs/heads/master"},"bb128fca0294f047d0071f4feb6564930c159d6a":{"kind":"REWORK","_number":2,"created":"2020-03-23 19:46:39.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/22/714522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/714522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/714522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/714522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/714522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/714522/2"}}},"commit":{"parents":[{"commit":"02df513aa8c6de4469c5d5214b1dd2591aeebe1a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02df513aa8c6de4469c5d5214b1dd2591aeebe1a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-23 19:45:52.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-23 19:46:34.000000000","tz":-300},"subject":"Soften the py3-only warning check","message":"Soften the py3-only warning check\n\nChange-Id: I4c770fab0366f70e7e01920498498a3b79bfc423\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb128fca0294f047d0071f4feb6564930c159d6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb128fca0294f047d0071f4feb6564930c159d6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
