)]}'
{"id":"openstack%2Frequirements~714834","triplet_id":"openstack%2Frequirements~master~I111fb609751b8724248a85fe9adf004ae3c4e634","project":"openstack/requirements","branch":"master","topic":"drop-pyscss2-again","hashtags":[],"change_id":"I111fb609751b8724248a85fe9adf004ae3c4e634","subject":"Drop pyscss2 and django-pyscss2","status":"MERGED","created":"2020-03-25 04:53:58.000000000","updated":"2020-03-25 14:12:31.000000000","submitted":"2020-03-25 14:10:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714834-1585145456967-4f1731f4","meta_rev_id":"e8c54b528a3cf18f60a8be9eb74fcbb33b07420c","_number":714834,"virtual_id_number":714834,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 14:10:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-03-25 06:38:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-03-25 12:46:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-03-25 12:46:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 06:38:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-25 12:46:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-25 14:10:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe5eb816dcb09be60e652262ab01233804a66684","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-25 04:53:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70a08a418e06155ac3b1f773d8dd50e21970356f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 06:18:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33ceb1121e942ad8b4fe55eb7db9c69 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4353227d6484f22b7874bf5f365562b : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff52dbe5aa2c4f10bada3ba3ed68f3a0 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46536800c458487aadc82acb7d594367 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2b8d953a67d4dc78695bca2183cf147 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c067ee70c0442a884f43551d1db75ad : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74236c3c1f0b4bb2a05633614a297b9c : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d59f264c7b314dd4a7079668fab8dccd : SUCCESS in 7m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c93a6543e7e74ead94dd55bdb3a720d7 : SUCCESS in 8m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1c6957dec97948c090faf081b38db83d : SUCCESS in 6m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5e107705a714f60b991aa3acc04259f : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b5266eb12114f20a73259b6e66d1cb4 : SUCCESS in 8m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/028516c7c1554bc29a1cff08e89a44c8 : SUCCESS in 12m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/69b3d6611f3d4bb49c62d01e56e98fc3 : SUCCESS in 29m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d9dc3cdc5cbb43dcb2d28ff23252faa3 : SUCCESS in 15m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/59d054e2b4d54e13a1be1602e09bd279 : SUCCESS in 12m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1328cbbe33a5478eb5ae12719c49e45f : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d215b8edb36a4e63923daefbc571730a : SUCCESS in 4m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bfebb734966a43b5ae815dd4c7536b3a : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/947741101f6d4554bf67344918d7f1af : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/57b3255b8fe0434dbe1f46c2ea0561fa : SUCCESS in 8m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4851316ff5b54b6ebd246ba664b01868 : SUCCESS in 8m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a096fc20de554d578a8c0b435ba89188 : SUCCESS in 55m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0f1f0c01db8482c9e041b499bf5ff44 : SUCCESS in 1h 04m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f0902216f3849aabf5e7178d53d3427 : SUCCESS in 1h 18m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/156594a4ad9f442b9c5325c9706fb306 : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":1},{"id":"87f8e70937f7a13a6a7fd43d988ca9bb8c972afa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-25 06:38:21.000000000","message":"Patch Set 1: Code-Review+2\n\nonly in requirements, not in use elsewhere in openstack (that I could find)","accounts_in_message":[],"_revision_number":1},{"id":"b62c2262fec58225e0681a37bd70b9deaefdff76","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-25 12:46:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTests look happy at least. Easy enough to revert if we find there\u0027s something that we don\u0027t have covered.","accounts_in_message":[],"_revision_number":1},{"id":"2de3d18466bef7d5aceb413ac33aac7cbc0e2ed6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 12:46:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e0d25c68b7cfa34df75dc502d3dff4ff76454a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 14:10:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9c62a5e3339532149e3087f5ebfd3882d8a920e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 14:10:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ffb66a132bd4532b662ea21025bb765 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfecaf9243cd41ec88d68526587e8c11 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5fc931ecba940c1a00f1f138dd31371 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36d86d859a004ccf81dcdd5e8085a596 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f65860cfbd6240948277e4c3e6c4da86 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8477dfea14934c2ea4705b79509f3551 : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/63912dd3049d49049187a12061b64f43 : SUCCESS in 8m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b391de3718ae4ed0a377d08ff0a98153 : SUCCESS in 9m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/92c80373904b4d909bbcdc0d12341cf6 : SUCCESS in 6m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7063d52530284c2faa21e1900aaf5f50 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eca3b084e8ba43fab1d5666ca9e2f1b4 : SUCCESS in 7m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7793634c702049dc832ca61611a3641b : SUCCESS in 12m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9fc3480c332f490bbcd53715255d754b : SUCCESS in 32m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3bcf8c15056245349c4b7cc8a1878871 : SUCCESS in 13m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7fdbc2edad414daf8c8bc6a8c386e42b : SUCCESS in 12m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d0dcaddb95644fb881ab5805d5457331 : SUCCESS in 16m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4266ac285efa4dd09332491cdfc65ec5 : SUCCESS in 4m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a91b054c97d84995a84a805cbe15ac51 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dcb092429ad54047a7922d73a72a7d01 : SUCCESS in 7m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3829ea5a609c4d638463683b84286708 : SUCCESS in 10m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1a9abb399f4c4b73942e68a68315b471 : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/280e130c009143ab8910e543996e37bb : SUCCESS in 55m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/668fcf589b0f494ab3af138c3163fce5 : SUCCESS in 1h 05m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b91f2233bdf48b7889b8ee8b007553a : SUCCESS in 1h 22m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cc751de3b1d2479dbd91908b24373da5 : SUCCESS in 1h 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9cfc261c71182db89127cefd6b9417b5f874d90c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 14:12:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0cb3df16966444a846c2d22c022896e : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ce4c4ca81d5588e5d935acbd9622f33625b13d4","revisions":{"4ce4c4ca81d5588e5d935acbd9622f33625b13d4":{"kind":"REWORK","_number":1,"created":"2020-03-25 04:53:58.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/34/714834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/714834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/714834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/714834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/714834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/714834/1"}}},"commit":{"parents":[{"commit":"06c73cd2a2628588a296cfe706aefedbfb2d2bbf","subject":"Merge \"update constraint for mistral-lib to new release 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06c73cd2a2628588a296cfe706aefedbfb2d2bbf"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-03-25 04:31:31.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-03-25 04:35:14.000000000","tz":540},"subject":"Drop pyscss2 and django-pyscss2","message":"Drop pyscss2 and django-pyscss2\n\npyscss is now maintained again and a new version commpatible with setuptools 46\nhas been released. Once a new horizon release is shipped, horizon plugins do not\nneed to install pyscss2 (and django-pyscss2), so pyscss2 and django-pyscss2\nwill be unnecessary.\n\nDepends-On: https://review.opendev.org/714826\nChange-Id: I111fb609751b8724248a85fe9adf004ae3c4e634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ce4c4ca81d5588e5d935acbd9622f33625b13d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ce4c4ca81d5588e5d935acbd9622f33625b13d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
