)]}'
{"id":"openstack%2Frequirements~714987","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I4269cdb44d216d719ecb7b05a6eedc2d9a043a63","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I4269cdb44d216d719ecb7b05a6eedc2d9a043a63","subject":"update constraint for os-brick to new release 2.10.2","status":"MERGED","created":"2020-03-25 16:29:38.000000000","updated":"2020-03-26 00:51:21.000000000","submitted":"2020-03-26 00:51:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714987-1585183881209-68d14c44","meta_rev_id":"22428ba4d1d7cf2660e01b3fb9f149dbb9694376","_number":714987,"virtual_id_number":714987,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-26 00:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-25 16:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-25 22:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-25 22:29:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 16:31:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-25 22:29:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-26 00:51:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4938c4c9947d86c0dbdacf736f05a5353bd14ed8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-25 16:29:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e665ce3d32cf771228641afc82cc5e7b35bd2ec","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-25 16:31:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ca2e37e930e0cb1a4dc7451ed946f9297be361ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 18:29:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/397d0d4f91fc4e82bde4624ee3d11f97 : FAILURE in 3m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c9e73e625424b569152f2426625fc37 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b4d6cd1f7374672b7d6e2c4e1e3ebae : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4e8ae79e9ea4de59a6106bf48feb32f : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a54a04f803cb4ca2a6dc4d9a68ea98a6 : SUCCESS in 6m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99d37f5014e64852b612a0733cc545b2 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44c9fa1b86df4d94895420c8e500bed8 : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d0b3dfd3f69e4966b96b136c96b876dc : SUCCESS in 7m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c28be1a67a94e149c096be98109eb3a : SUCCESS in 8m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0a9ce8890a934d5faf07e21e4bb23273 : SUCCESS in 12m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7683fddf52044c32ba1d1d427ccfce00 : SUCCESS in 11m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f02b181007dc41298925e5447c6d957a : SUCCESS in 5m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e0bbcd1de24d405cb06b43354e18b511 : SUCCESS in 7m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0f65dc9152624afd8a92179534d703ba : SUCCESS in 10m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7a0a3ee3a47343b8879f54e6bf6c37d6 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f6a60e4e2ca4477b5bd8a35ac32f346 : SUCCESS in 7m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/433bad60a81f49b9a0fc50b844fdce14 : SUCCESS in 23m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0ba2b934c74b458dacda5a5a3287dc92 : SUCCESS in 16m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9087a0afeb44451683eda35146a650d0 : SUCCESS in 24m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6ce907f1394b4eec9074c1fbaa489997 : SUCCESS in 21m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9070e963148c47efa7ad986f7e496f77 : SUCCESS in 13m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/88a0ebeffac34eafa3d91d6767c0fb9d : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6d58838f018c4f5193a1a461401ae127 : SUCCESS in 13m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5f09ac441a540fb9e92bd22f289c819 : SUCCESS in 21m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4640a59f67b84a669ff799bddb75b497 : SUCCESS in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3e4ff7eb07464c9a8d869c63b8bafa8c : SUCCESS in 9m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e15d5799b146454486f0bab1486191cf : SUCCESS in 10m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d717cbd65b9c4c99a1bb9ee4440ce2e2 : SUCCESS in 1h 13m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3e234ef8c8a24b8882015c82759a38d3 : SUCCESS in 1h 56m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/10e99a6344384954b768dab6aaf6b85e : FAILURE in 1h 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"aba0f00564710ac6ec27a7055c7b52266bd95299","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-25 18:37:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa26772840cf01600cd75de7583a7d8f483a39b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 20:55:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e8b85cf1b0204e2c8429469e15d53169 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3a339a06894cd7b5235b949dd29063 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0829dff981ce42629fad500075025026 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad1f0219cd404bc39f10834b4354e194 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efcbb91d5aaf4ce9b48d13d6076eb53b : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6883a1ba9e8b462189126520be72c1de : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b3448e2d430844999620a377cdde73a2 : SUCCESS in 5m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5b2155ccb2b049e08bb114aa9b457ce3 : SUCCESS in 9m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f01ca73c50ad4b6089a7535ce701fa84 : SUCCESS in 9m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4172bf2be25e4bbd9429fe62f4915164 : SUCCESS in 10m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8623393d879f4d23ae1276cf94e425dc : SUCCESS in 10m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e895a1454bff4dabac06e60f3091c5bd : SUCCESS in 7m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/42ea0bcdb5064444955aedd41334c9ce : SUCCESS in 6m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0230f6e68eed4c9c8eb2116e4233ed66 : SUCCESS in 12m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d70c3e78a0dc442395d4b449c8e45c0c : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a893eabaaae14f0fa23136970d5540b5 : SUCCESS in 8m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/da27026c2d7745ac85c8c134366ff8e6 : SUCCESS in 21m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b274e45855ad4c4c9a543b9a639c1902 : SUCCESS in 18m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a14ddaa1d1b04b7cb7a2602e0c9c204f : SUCCESS in 24m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b9251c0e401a47ceaea109151005aa46 : SUCCESS in 24m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/72adf503459a4436afe4ff1f353ff93b : SUCCESS in 14m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b6509f88e9247bba957004c421eff8a : SUCCESS in 15m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7af97c179f0e4a419b75060b72430920 : SUCCESS in 16m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a9116ccf6aee4755af2d773a5c811e7c : SUCCESS in 22m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee5968ebb2a342c5b872cb68ecda9f25 : SUCCESS in 8m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/401cd5ab07254b11b32eaec7dd2b0112 : SUCCESS in 21m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/28ad4e1d046e401fbb5c7fb7eb5d7d73 : SUCCESS in 11m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd411ab884024bec86c57245281b33d3 : SUCCESS in 1h 05m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/234d3ee5e17c44afb4e3baf40d85a774 : SUCCESS in 1h 27m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c270114ca04f47a1bba10cac6f312bf9 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0dccadccc4e19ca9319b93ba33c09e99c3d380c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-25 22:29:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d6788339040b3ffbfd936ca6bc6ed7619c6ea94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 22:30:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"059b12108851e2b16c2163512e449b008cedc408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 00:51:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ac795795324afa8fc6772ab8915b6a : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6725d1a9ed17451eaf8b6f3db14d99e0 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe932f243aaa488588dd309e82a7d347 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ded692567404d0bbaf5019b9e89cae6 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c61d84063b06416f95b031cb6abde458 : SUCCESS in 7m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/31dc0f7da9bf442ba27a885388623f79 : SUCCESS in 5m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e15c03e8949f45d2b72ef914eed0d188 : SUCCESS in 11m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b97b0bb87ae94bcdaee0a5656ae90dd3 : SUCCESS in 13m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6d920532d3304a05b3bc0ea9269af0b4 : SUCCESS in 15m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fb70147a72994efe9b7b12e799356bff : SUCCESS in 12m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d6985f59b96c41c394c73d17feb937a4 : SUCCESS in 12m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5cd7e4a5e39e4661b53a76e5d70b25bb : SUCCESS in 8m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a0afd3db0e3a4796a987c6e463359462 : SUCCESS in 12m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3ca6274c2a004c3ea6794fe42eb8f2d3 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2c956cf4ec14ae8bb201661df4a6599 : SUCCESS in 7m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/133ecd5f4aeb47f09e7b76805fae6ece : SUCCESS in 23m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9748b9592f71407fb61c93abc7942c41 : SUCCESS in 25m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/07ab7e460091431ca6594e842f648761 : SUCCESS in 22m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c22167c94c7146749cdea65c1d564341 : SUCCESS in 25m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3949a9abe73942df9cdcd2b3c495144a : SUCCESS in 21m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1948105b65df488987037599a9e8b230 : SUCCESS in 15m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7baf78db5c5a47628e9062f78677f3d0 : SUCCESS in 14m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2d974e6770341b7be9a510df8fd569f : SUCCESS in 22m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73dd19590d7e4b87b71a6fb48d3d9b98 : SUCCESS in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/de73f51295e3449098520e7eff581f47 : SUCCESS in 18m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8ae87d75b1864d7ea8619483bbeb8f95 : SUCCESS in 11m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5559e2e227e4e06a8ac4cb9eeafbae2 : SUCCESS in 1h 14m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/64e225d81c294fc381e27f7e1d3de373 : SUCCESS in 1h 53m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1b900c6e56f54e049375e062208567b4 : SUCCESS in 1h 19m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2c0b78c0cb4cabb99228c9e92fe51cfc3a2c496f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 00:51:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63f906acd2a295b21a14b6f5c0af8d31f78570a6","revisions":{"63f906acd2a295b21a14b6f5c0af8d31f78570a6":{"kind":"REWORK","_number":1,"created":"2020-03-25 16:29:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/714987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/714987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/714987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/714987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/714987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/714987/1"}}},"commit":{"parents":[{"commit":"3410c3d2afa4e607012b4c09d191575ceebb4438","subject":"update constraint for openstacksdk to new release 0.36.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3410c3d2afa4e607012b4c09d191575ceebb4438"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-25 16:29:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-25 16:29:37.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.10.2","message":"update constraint for os-brick to new release 2.10.2\n\nChange-Id: I4269cdb44d216d719ecb7b05a6eedc2d9a043a63\nmeta:version: 2.10.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: If7618a528a88fe92e96c118cd771d904a6f4f13a\nmeta:release:Code-Review+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63f906acd2a295b21a14b6f5c0af8d31f78570a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63f906acd2a295b21a14b6f5c0af8d31f78570a6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
