)]}'
{"id":"openstack%2Frequirements~715022","triplet_id":"openstack%2Frequirements~stable%2Fstein~I72c03c06de0f1e029d32f46446fb18f0ae25604a","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I72c03c06de0f1e029d32f46446fb18f0ae25604a","subject":"update constraint for os-brick to new release 2.8.4","status":"MERGED","created":"2020-03-25 18:27:59.000000000","updated":"2020-03-26 00:51:22.000000000","submitted":"2020-03-26 00:51:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715022-1585183882675-dfe5b5ea","meta_rev_id":"d3d7e7ee6f8f58b75e63e1f67df95ab8fcb5bca2","_number":715022,"virtual_id_number":715022,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-26 00:51:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-25 18:37:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-25 22:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-25 22:29:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 18:37:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-25 22:29:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-26 00:51:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0f52e98ff72fbfcae62dbb8a0871d4b09c8781c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-25 18:27:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acd085e0dbe68f8a61aef5b06ac77b38d78dd8ad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-25 18:37:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df116abef4c41a2d3dda93901a01af6710b12132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:14:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f6189252439b4dcd924e3e7e96b8ca79 : FAILURE in 5m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5982c3122624d53b3e84ec40798f931 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df07b32bd3904e3381c0122ced60ec67 : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/16dc0f32427a4a9c82a0b3b61aac09c4 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4cbab8ff70141bcaafd8e223c7860f9 : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6eae9dd278443baa741ffa7679fe40f : SUCCESS in 6m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a86f099baed34d2da099b3c06577421f : SUCCESS in 6m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a6db1affb2264154aa3b90eacca6ca0f : SUCCESS in 9m 48s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/82b9ff9f6f9e4e328ce6dcf71a071860 : SUCCESS in 10m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/03d506b7408e45feb558354144495a5c : SUCCESS in 9m 29s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/308cf13617b544638dc4ec490fe95b21 : SUCCESS in 9m 51s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/dda118da4592445b950d93b4e0e31feb : SUCCESS in 11m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4c0b30758b3b45d4b041c3b1b2d8d159 : SUCCESS in 7m 12s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/84a96e8a83e64e39bc1e2915f73f2da9 : SUCCESS in 7m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2085fe2f79f64f889147aff7d23d717f : SUCCESS in 10m 47s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/0853a058c523401499fc1363ab8420c2 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/310c2d4b35df49c9b467216de8fc5ed7 : SUCCESS in 10m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5a15b2d9460745af869f314812bdb726 : SUCCESS in 23m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/032276bdfdcd4200b75fb6b0017e52b3 : SUCCESS in 29m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/76a651266b1842a1b78706ef1c470dbd : SUCCESS in 21m 30s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/faf80766d03c413f8d5b14e3e0968ed9 : SUCCESS in 25m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/af1af901c52f497eb26846ef505cc0eb : SUCCESS in 16m 40s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f42165a6549440b78c7ddff738d6fae7 : SUCCESS in 14m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c540df702330400a8a5fdadf87362277 : SUCCESS in 20m 16s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/ec9a5a2c8b2e43b2a59b875c6fdc8f84 : SUCCESS in 7m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ef57764b19654f65b0a3496b59a10d17 : SUCCESS in 14m 56s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/ef11e6d052304d7b958336f467f12608 : SUCCESS in 11m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb9802de1fbd4c38ad327c9b23315464 : SUCCESS in 1h 20m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7110c5d4749645f289c018635a676e73 : SUCCESS in 1h 45m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b98f55d050594cdab1ffa35f96e34c9a : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c074346e4200bcc68d370c1e7de0a5528c5be469","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-25 22:29:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4fda19c5739c4f5e1dd9fd82a285f8bf56a1c543","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 22:30:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be537bad9ec3751d31967a9fe38e87fbb9a51f1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 00:51:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/129de73cb23c4d26a848c1c2cbadfea9 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b42cbb8c13a24e0787b88de78182c732 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f6dc554bf7a04777998d6a16c0fa2800 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5e7866a1ffa401eb20f48149a38b113 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/beebca9f71244888965b4dbbaec38313 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6199473fd53a4072afdd6f07c0585c64 : SUCCESS in 4m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fb4635e437404f318420e688f5a9e220 : SUCCESS in 8m 47s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/8c4044d2a7984fecba11fe97adf6b382 : SUCCESS in 8m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8b75fa8603314770929d0e29807ded1d : SUCCESS in 10m 42s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/31cb8d821c744f169a232a20a4404681 : SUCCESS in 9m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3301387660a149feb3d968ef61501621 : SUCCESS in 8m 02s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/9d50ea5bb96c486aa7569c8846fb406d : SUCCESS in 6m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/706b711c283b4a7aa6255a73271c7f0c : SUCCESS in 12m 13s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/5eec72fcc9ac482ab738d1691611fa1f : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0ce27c68894407fbe2510baf7e25902 : SUCCESS in 10m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/64f70f1f301f4afc9421e1b6bf28712f : SUCCESS in 23m 22s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/594ad3d85aff456494db60ce96145fce : SUCCESS in 28m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0a97d86356c14d8fb49e7c2e6b4e5303 : SUCCESS in 21m 58s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ae519ebc0a544382b0435b81b750f8dd : SUCCESS in 24m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b1781f3bd51b4b51900b2b34000d215b : SUCCESS in 13m 31s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9276a77b89d746d3abe1b1732a8f25c0 : SUCCESS in 15m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/541987387c44465d93666587865482b1 : SUCCESS in 19m 25s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/02bc2b5e664e4ba599a43198f99f4097 : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2356e38b77cf438a8ce99773e094d132 : SUCCESS in 8m 40s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/489065022f5e45ef922f295561baa7df : SUCCESS in 8m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a79aeff85c1246d28ee2e470d4b314e9 : SUCCESS in 1h 09m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d3786bd8b4094a9d8a7eeddbda08e5a9 : SUCCESS in 1h 27m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c442810e83b54981b8aa5383bdfae900 : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3b55bb04d0331064269cb60b41085cd15a119b70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 00:51:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12fedb96508e38b08203605373d451ceaf0dd2e2","revisions":{"12fedb96508e38b08203605373d451ceaf0dd2e2":{"kind":"REWORK","_number":1,"created":"2020-03-25 18:27:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/715022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/715022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/715022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/715022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/715022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/715022/1"}}},"commit":{"parents":[{"commit":"7cead2788f32daad813236627dcb9b6862e10645","subject":"Merge \"update constraint for keystonemiddleware to new release 6.0.1\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7cead2788f32daad813236627dcb9b6862e10645"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-25 18:27:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-25 18:27:57.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.8.4","message":"update constraint for os-brick to new release 2.8.4\n\nChange-Id: I72c03c06de0f1e029d32f46446fb18f0ae25604a\nmeta:version: 2.8.4\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I875a54464bb2173c2fc1fc3c8940e2822b7df4be\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12fedb96508e38b08203605373d451ceaf0dd2e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12fedb96508e38b08203605373d451ceaf0dd2e2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
