)]}'
{"id":"openstack%2Frequirements~715199","triplet_id":"openstack%2Frequirements~master~I8e65a12c1a9861d08667f7828a4c2c024d3a4605","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8e65a12c1a9861d08667f7828a4c2c024d3a4605","subject":"update constraint for openstacksdk to new release 0.44.0","status":"ABANDONED","created":"2020-03-26 13:10:09.000000000","updated":"2020-04-07 11:08:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8262da2d1cb2d96dd5c6ef87a5ff2108798d89d5","_number":715199,"virtual_id_number":715199,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-26 14:57:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"date":"2020-03-26 15:32:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-2,"date":"2020-03-26 15:29:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-26 14:57:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-26 15:29:40.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-03-26 15:32:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"e6c43089838579dbd0bccc528025096da93f2145","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-26 13:10:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15ddd3c8c11698a07dd4d3cc4c0f419b5dfd7a4b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-26 13:37:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1004f4ec0b7810c340a0ffb6508a74bdb01a3449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 14:57:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f54f7f4e54743289ece148efc6d2f79 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffa7c7e94b044cc187d6dbf12e6e6992 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad577bb14b1247eba4e6d63d7fb66c4b : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a37747c505ba435bb4734b33ea6a1328 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/680c00a1f34245dcbb0e566ff224d8d8 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c912b1fdd0a4a6ca34d45193d7266b0 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04e9e69b38d9404d8eeed0a03d8314c0 : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9184d99190994652802c22d90baf7ce7 : FAILURE in 14m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b14ed83a30d74f3aab79f8078dbab771 : FAILURE in 7m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/03c013167c4f4477a0f78c54ba03174c : FAILURE in 4m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/065298abee364fa4932b3ea250b04a1b : FAILURE in 4m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ef2f5bb79da45de8a12d02c0cd06c90 : SUCCESS in 9m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e73fcd4563144bdb30e81782c7a50e9 : FAILURE in 4m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f7f9cc97b0654952b276cc4f21801cbe : SUCCESS in 30m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e57cb9d630c24ef181a4a5dc991d49bc : SUCCESS in 17m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1c63685bd6054fab92e968dcefaba3a5 : SUCCESS in 15m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e5cff7c6df049d88bad0a70b83328af : SUCCESS in 18m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4ba84fb44e7f4ab4868d9dad24f91767 : FAILURE in 6m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6ef9f14d020c46d7b55b1b58e5f4e0ed : FAILURE in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/125e263f30264639826e3a9ac07e4a4b : FAILURE in 4m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0b6971a5242446c396fc55b9e3f55d11 : FAILURE in 5m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/443b714e053f4d8498d1c90644156e25 : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b2b2e041ebf43719561158870935f1a : SUCCESS in 1h 05m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/268035e90e2e496c950e340357dbba2a : SUCCESS in 1h 15m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9554883accaa4d17a9cbbc41f9090aea : FAILURE in 13m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cec48bf2e0f04e8aa5543b261928da67 : FAILURE in 51m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6d79d2af7df9350b669fd831b4c44228dd6e0bd6","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-26 15:29:40.000000000","message":"Patch Set 1: Code-Review-2\n\nBlock this. We cut a bad release and would prefer nothing in openstack use it.","accounts_in_message":[],"_revision_number":1},{"id":"6169bf4111f6c69957e545fb3b2d6bf62955681f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-26 15:32:17.000000000","message":"Patch Set 1: -Code-Review\n\nOne thing I missed with voting is if we did this, or a follow up, we need to update the bot proposed patch to break out the versions based on python_version. So it should be fine for openstack projects, but doesn\u0027t address external consumers.","accounts_in_message":[],"_revision_number":1},{"id":"3a36e9563fdafb61fc7ecf8275e613b5dfd4c456","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 11:08:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef3bf98cb7a0a30b5dd86f80f8204b0037468d4a","revisions":{"ef3bf98cb7a0a30b5dd86f80f8204b0037468d4a":{"kind":"REWORK","_number":1,"created":"2020-03-26 13:10:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/715199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/715199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/715199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/715199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/715199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/715199/1"}}},"commit":{"parents":[{"commit":"fe8ad8b87b8d7ebcba11d9c4c97638ba006cd0ca","subject":"Merge \"update constraint for oslo.cache to new release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe8ad8b87b8d7ebcba11d9c4c97638ba006cd0ca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-26 13:10:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-26 13:10:07.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.44.0","message":"update constraint for openstacksdk to new release 0.44.0\n\nChange-Id: I8e65a12c1a9861d08667f7828a4c2c024d3a4605\nmeta:version: 0.44.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: I48968259159fdd12684e41434de1e588de9b7eac\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef3bf98cb7a0a30b5dd86f80f8204b0037468d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef3bf98cb7a0a30b5dd86f80f8204b0037468d4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
