)]}'
{"id":"openstack%2Frequirements~715518","triplet_id":"openstack%2Frequirements~master~I496973e204b4f486ba3ce13faed545faf6392762","project":"openstack/requirements","branch":"master","topic":"python-requires","hashtags":[],"change_id":"I496973e204b4f486ba3ce13faed545faf6392762","subject":"Replace old requires-python with python-requires","status":"MERGED","created":"2020-03-27 18:47:54.000000000","updated":"2020-03-28 00:54:01.000000000","submitted":"2020-03-28 00:52:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715518-1585356749869-1fb5c97b","meta_rev_id":"c6f8ed53f9f404511923f59bf5af68f2012df20d","_number":715518,"virtual_id_number":715518,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-28 00:52:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-27 18:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-27 23:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-27 23:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 18:48:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-27 23:20:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-28 00:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62254898d49d6b9bcc19aca6256663d90030bef3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-27 18:47:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d91a44c5b76e7e8443d62d364e21c7a79c7828bf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-27 18:48:30.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"70365deda1dd71c438f8c1d0c047ac7c87800437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 20:27:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e28bfe747f954b018ce63a19c0cedf05 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a8977b47fde460f8628fcd05e114d57 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b244e23d385485a9d723c6e40a884f0 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/922fcbbc6b1e40778bc3aebb0c2799d0 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e622f0fcfe4840f39e7109c33fd68e8d : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea1829860ff44858a2b83a41056baf92 : SUCCESS in 4m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9897dfc004474a239c729de015e040c1 : SUCCESS in 55m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04ca9611c8aa4898b244689f24f29fd5 : SUCCESS in 1h 05m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e3066fd9bcc483da9ccf8b825b7c1ce : SUCCESS in 1h 33m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b0a785d0f7244980b996827a505258ef : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":1},{"id":"fa70f1d5dfa79015e0fcfa4775df68abb95be8ff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-27 23:20:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be4d90ddff6c6d511ece32ef3efc3a1188039106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 23:21:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d4fcba14bcb4a83c9505ba44cd39c9dcc8c8244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 00:52:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce568d4beb04cdab1d3fcc54c5031f8 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0982c6f049d4722850f56d1ba246af9 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f98186fd83724730848d3f7fe865806f : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58ff13d363b144a3b3a8c1d4fbd79215 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/065cef39212642d68f8de60d8b6085cd : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/daf870cfaadc480d97ab8e576562f909 : SUCCESS in 53m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a026b4dce64141138fbc673225347b10 : SUCCESS in 1h 16m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec7391fc0a3d4e2a9e7d2f832d0524ab : SUCCESS in 1h 28m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cc4ab2440ef84f4595a80532b69de85c : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":1},{"id":"441c20b368e3590d06dc6ff214276eff1315fde0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 00:52:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4d73056bfd89e52f4b9decb7343e8431f82bd38c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 00:54:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d48088bc8754c048b722666ac73c303 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15b2e73c0360aeca3284e8701a5dc8e3e37b2fb0","revisions":{"15b2e73c0360aeca3284e8701a5dc8e3e37b2fb0":{"kind":"REWORK","_number":1,"created":"2020-03-27 18:47:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/715518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/715518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/715518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/715518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/715518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/715518/1"}}},"commit":{"parents":[{"commit":"3ee6f139f5dc9c202dd66693ab1ed88e328e147f","subject":"Merge \"update constraint for XStatic-Bootstrap-SCSS to new release 3.4.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ee6f139f5dc9c202dd66693ab1ed88e328e147f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-27 18:47:38.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-27 18:47:38.000000000","tz":-300},"subject":"Replace old requires-python with python-requires","message":"Replace old requires-python with python-requires\n\nrequires-python is old value and does not work, python-requires\nis right metadata.\n\n\u0027python-requires\u0027 is introduced from PEP440 and requires-python was old value\nin PEP345.\n- https://packaging.python.org/guides/distributing-packages-using-setuptools/#python-requires\n- https://www.python.org/dev/peps/pep-0345/#requires-python\n\nChange-Id: I496973e204b4f486ba3ce13faed545faf6392762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15b2e73c0360aeca3284e8701a5dc8e3e37b2fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15b2e73c0360aeca3284e8701a5dc8e3e37b2fb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
